Skip to content

fix: Admin can't change room name - #31713

Merged
kodiakhq[bot] merged 5 commits into
developfrom
fix/roomName-change
Feb 15, 2024
Merged

fix: Admin can't change room name#31713
kodiakhq[bot] merged 5 commits into
developfrom
fix/roomName-change

Conversation

@dougfabris

@dougfabris dougfabris commented Feb 9, 2024

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Closes #31722

Steps to test or reproduce

Further comments

SUP-490

@changeset-bot

changeset-bot Bot commented Feb 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b2a3f69

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Feb 9, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (100585f) 54.38% compared to head (b2a3f69) 54.38%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31713      +/-   ##
===========================================
- Coverage    54.38%   54.38%   -0.01%     
===========================================
  Files         2273     2272       -1     
  Lines        50085    50074      -11     
  Branches     10211    10212       +1     
===========================================
- Hits         27237    27231       -6     
+ Misses       20371    20364       -7     
- Partials      2477     2479       +2     
Flag Coverage Δ
e2e 53.17% <100.00%> (-0.04%) ⬇️
e2e-api 39.96% <ø> (+0.03%) ⬆️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@scuciatto scuciatto added this to the 6.7 milestone Feb 9, 2024
@dougfabris
dougfabris marked this pull request as ready for review February 9, 2024 19:50
@dougfabris
dougfabris requested a review from a team as a code owner February 9, 2024 19:50
gabriellsh
gabriellsh previously approved these changes Feb 9, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Feb 15, 2024
@dougfabris dougfabris removed this from the 6.7 milestone Feb 15, 2024
@kodiakhq
kodiakhq Bot merged commit 8ed3720 into develop Feb 15, 2024
@kodiakhq
kodiakhq Bot deleted the fix/roomName-change branch February 15, 2024 15:05
@dougfabris dougfabris added this to the 6.7 milestone Apr 3, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Unable to change room names

3 participants