Skip to content

fix: Allow only numbers, if trigger's condition is 'visitor time on site' - #31666

Merged
kodiakhq[bot] merged 4 commits into
developfrom
fix/triggerConditionInput
Feb 7, 2024
Merged

fix: Allow only numbers, if trigger's condition is 'visitor time on site'#31666
kodiakhq[bot] merged 4 commits into
developfrom
fix/triggerConditionInput

Conversation

@dougfabris

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Fixes #21702

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Feb 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f1e3329

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dougfabris
dougfabris marked this pull request as ready for review February 6, 2024 19:06
@dougfabris
dougfabris requested a review from a team as a code owner February 6, 2024 19:06
MartinSchoeler
MartinSchoeler previously approved these changes Feb 6, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Feb 6, 2024
@codecov

codecov Bot commented Feb 6, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (dbaf87f) 54.41% compared to head (f1e3329) 54.36%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31666      +/-   ##
===========================================
- Coverage    54.41%   54.36%   -0.06%     
===========================================
  Files         2280     2280              
  Lines        50155    50155              
  Branches     10232    10232              
===========================================
- Hits         27294    27267      -27     
- Misses       20383    20402      +19     
- Partials      2478     2486       +8     
Flag Coverage Δ
e2e 53.17% <ø> (-0.10%) ⬇️
e2e-api 39.94% <ø> (ø)
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Feb 7, 2024
@kodiakhq
kodiakhq Bot merged commit 789544f into develop Feb 7, 2024
@kodiakhq
kodiakhq Bot deleted the fix/triggerConditionInput branch February 7, 2024 13:57
@dougfabris dougfabris added this to the 6.7 milestone Apr 3, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Input number validation

2 participants