Skip to content

feat: Improve 2fa modals and change resend email to button - #31621

Merged
kodiakhq[bot] merged 10 commits into
developfrom
chore/change-2fa-modal
Feb 6, 2024
Merged

feat: Improve 2fa modals and change resend email to button#31621
kodiakhq[bot] merged 10 commits into
developfrom
chore/change-2fa-modal

Conversation

@rique223

@rique223 rique223 commented Feb 2, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Improved the 2fa email and TOTP modals by:

  • Changing their copies to improve the language and message of the text;
  • Adding a tagline to both of them;
  • Removing the icon from both of them;
  • Changing the email modal 'Resend email' link to a small button;
  • Changing the input placeholder of both of them;

Demo images:

Email modal:
image

TOTP modal:
image

Issue(s)

Steps to test or reproduce

Further comments

WM-223

Changed the copy of the 2fa modal, added a tagline, changed the resend email link to a button and removed the icon.
Changed the copy, removed the icon and added a tagline to the 2fa TOTP modal.
@changeset-bot

changeset-bot Bot commented Feb 2, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f3fc767

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Feb 2, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (f612d74) 54.39% compared to head (f3fc767) 54.36%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31621      +/-   ##
===========================================
- Coverage    54.39%   54.36%   -0.03%     
===========================================
  Files         2276     2276              
  Lines        50105    50105              
  Branches     10227    10227              
===========================================
- Hits         27253    27240      -13     
- Misses       20369    20389      +20     
+ Partials      2483     2476       -7     
Flag Coverage Δ
e2e 53.23% <ø> (+0.03%) ⬆️
e2e-api 39.83% <ø> (-0.15%) ⬇️

Flags with carried forward coverage won't be shown. Click here to find out more.

@rique223
rique223 marked this pull request as ready for review February 2, 2024 19:08
@rique223
rique223 requested a review from a team as a code owner February 2, 2024 19:08

@hugocostadev hugocostadev left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

fix conflicts and add changeset, thanks! 🚀

@rique223 rique223 changed the title refactor: Improve 2fa modals feat: Improve 2fa modals and change resend email to button Feb 6, 2024
hugocostadev
hugocostadev previously approved these changes Feb 6, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Feb 6, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Feb 6, 2024
@kodiakhq
kodiakhq Bot merged commit 46e8272 into develop Feb 6, 2024
@kodiakhq
kodiakhq Bot deleted the chore/change-2fa-modal branch February 6, 2024 21:50
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants