Skip to content

fix: ldap avatar sync background process never runs - #31604

Merged
kodiakhq[bot] merged 3 commits into
developfrom
fix/ldap-background-sync
Feb 14, 2024
Merged

fix: ldap avatar sync background process never runs#31604
kodiakhq[bot] merged 3 commits into
developfrom
fix/ldap-background-sync

Conversation

@pierre-lehnen-rc

@pierre-lehnen-rc pierre-lehnen-rc commented Feb 1, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Issue(s)

SUP-476

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Feb 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1b24fd3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Feb 1, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (7a1cf2a) 54.32% compared to head (1b24fd3) 54.35%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31604      +/-   ##
===========================================
+ Coverage    54.32%   54.35%   +0.02%     
===========================================
  Files         2272     2271       -1     
  Lines        50075    50044      -31     
  Branches     10213    10362     +149     
===========================================
- Hits         27204    27200       -4     
+ Misses       20390    20357      -33     
- Partials      2481     2487       +6     
Flag Coverage Δ
e2e 53.18% <ø> (+0.03%) ⬆️
e2e-api 39.91% <ø> (+0.05%) ⬆️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@pierre-lehnen-rc
pierre-lehnen-rc marked this pull request as ready for review February 1, 2024 17:07
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Feb 14, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Feb 14, 2024
@kodiakhq
kodiakhq Bot merged commit be8cd9b into develop Feb 14, 2024
@kodiakhq
kodiakhq Bot deleted the fix/ldap-background-sync branch February 14, 2024 20:15
@scuciatto scuciatto added this to the 6.7 milestone Feb 15, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants