fix: missing slashcommand permissions - #31594
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 9c4d854 The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## release-6.6.0 #31594 +/- ##
==================================================
+ Coverage 49.54% 59.87% +10.33%
==================================================
Files 3322 1796 -1526
Lines 81748 35037 -46711
Branches 16874 7302 -9572
==================================================
- Hits 40501 20980 -19521
+ Misses 36509 12454 -24055
+ Partials 4738 1603 -3135
Flags with carried forward coverage won't be shown. Click here to find out more. |
ggazzo
force-pushed
the
fix/slash-permission
branch
from
February 1, 2024 02:33
5d788ad to
0c12724
Compare
ggazzo
force-pushed
the
fix/slash-permission
branch
from
February 1, 2024 02:36
0c12724 to
f8297a3
Compare
ggazzo
marked this pull request as ready for review
February 1, 2024 11:54
sampaiodiego
requested changes
Feb 1, 2024
Co-authored-by: Diego Sampaio <chinello@gmail.com>
sampaiodiego
previously approved these changes
Feb 1, 2024
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments