fix: Favoriting room through rooms page - #31554
Merged
kodiakhq[bot] merged 13 commits intoFeb 5, 2024
Merged
Conversation
🦋 Changeset detectedLatest commit: 8fa6543 The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## develop #31554 +/- ##
==========================================
Coverage ? 46.18%
==========================================
Files ? 2924
Lines ? 75410
Branches ? 15432
==========================================
Hits ? 34828
Misses ? 36316
Partials ? 4266
Flags with carried forward coverage won't be shown. Click here to find out more. |
Contributor
Author
Contributor
Contributor
Author
Member
|
Hello, thanks for reporting this issue! This |
Fixed a bug where the favorite field in the save room settings payload was being overridden by the favorite field returned from the getDirtyFields function. Consequently, this resulted in always sending the favorite field as undefined to the back-end. Deleting the field before building the payload solved the issue.
…Rocket.Chat into favorite-setting-functioning
hugocostadev
suggested changes
Jan 30, 2024
hugocostadev
previously approved these changes
Jan 30, 2024
hugocostadev
approved these changes
Feb 5, 2024
sampaiodiego
pushed a commit
that referenced
this pull request
Feb 17, 2024
Co-authored-by: Henrique Guimarães Ribeiro <43561537+rique223@users.noreply.github.com>
Merged
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.




Proposed changes
Fixed a bug where the favorite field in the save room settings payload was being overridden by the favorite field returned from the getDirtyFields function. Consequently, this resulted in always sending the favorite field as undefined to the back-end. Deleting the field before building the payload solved the issue.
Demo gif:

Issue(s)
Closes #31545
Steps to test or reproduce
Introduced here: #28318
WM-225