feat: Message keyboard navigability - #31549
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 5b5396a The changes in this PR will be included in the next version bump. This PR includes changesets to release 31 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAttention:
Additional details and impacted files@@ Coverage Diff @@
## develop #31549 +/- ##
===========================================
+ Coverage 54.37% 54.40% +0.02%
===========================================
Files 2273 2273
Lines 50077 50098 +21
Branches 10212 10222 +10
===========================================
+ Hits 27230 27255 +25
+ Misses 20366 20361 -5
- Partials 2481 2482 +1
Flags with carried forward coverage won't be shown. Click here to find out more. |
ggazzo
reviewed
Feb 15, 2024
dougfabris
marked this pull request as ready for review
February 16, 2024 15:43
dougfabris
force-pushed
the
feat/message-keyboard-navigation
branch
from
February 19, 2024 17:33
67f7940 to
1b59c1e
Compare
ggazzo
approved these changes
Feb 20, 2024
Merged
This was referenced Jul 25, 2025
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Closes #31300
Steps to test or reproduce
Further comments
DSAR-428
DSAR-450