Skip to content

fix: UI crashing when deleting room - #31546

Merged
kodiakhq[bot] merged 3 commits into
developfrom
fix/deleteRoom
Jan 26, 2024
Merged

fix: UI crashing when deleting room#31546
kodiakhq[bot] merged 3 commits into
developfrom
fix/deleteRoom

Conversation

@gabriellsh

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@gabriellsh
gabriellsh requested a review from a team as a code owner January 26, 2024 14:16
@changeset-bot

changeset-bot Bot commented Jan 26, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b666ffb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 26, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (a200538) 49.54% compared to head (b666ffb) 49.55%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31546      +/-   ##
===========================================
+ Coverage    49.54%   49.55%   +0.01%     
===========================================
  Files         3329     3328       -1     
  Lines        81802    81771      -31     
  Branches     16883    16877       -6     
===========================================
- Hits         40529    40523       -6     
+ Misses       36520    36500      -20     
+ Partials      4753     4748       -5     
Flag Coverage Δ
e2e 53.30% <100.00%> (+0.02%) ⬆️
e2e-api 40.62% <ø> (+<0.01%) ⬆️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dougfabris dougfabris left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@gabriellsh gabriellsh added this to the 6.6 milestone Jan 26, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 26, 2024
@kodiakhq
kodiakhq Bot merged commit 62bd278 into develop Jan 26, 2024
@kodiakhq
kodiakhq Bot deleted the fix/deleteRoom branch January 26, 2024 17:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants