Skip to content

chore: Add missing changeset - #31461

Merged
kodiakhq[bot] merged 1 commit into
developfrom
chore/missing-saml-changeset
Jan 16, 2024
Merged

chore: Add missing changeset#31461
kodiakhq[bot] merged 1 commit into
developfrom
chore/missing-saml-changeset

Conversation

@matheusbsilva137

@matheusbsilva137 matheusbsilva137 commented Jan 16, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Related to #28366

Issue(s)

Steps to test or reproduce

Further comments

SUP-430

@changeset-bot

changeset-bot Bot commented Jan 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 21f569b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 16, 2024
@codecov

codecov Bot commented Jan 16, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (6526b2f) 49.57% compared to head (21f569b) 49.58%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31461      +/-   ##
===========================================
+ Coverage    49.57%   49.58%   +0.01%     
===========================================
  Files         3316     3316              
  Lines        81517    81517              
  Branches     16718    16718              
===========================================
+ Hits         40415    40424       +9     
+ Misses       36394    36390       -4     
+ Partials      4708     4703       -5     
Flag Coverage Δ
e2e 53.19% <ø> (+<0.01%) ⬆️
e2e-api 40.70% <ø> (+0.01%) ⬆️
unit 76.51% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@matheusbsilva137 matheusbsilva137 added this to the 6.6 milestone Jan 16, 2024
@matheusbsilva137
matheusbsilva137 marked this pull request as ready for review January 16, 2024 16:28
@kodiakhq
kodiakhq Bot merged commit 224e194 into develop Jan 16, 2024
@kodiakhq
kodiakhq Bot deleted the chore/missing-saml-changeset branch January 16, 2024 16:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants