Skip to content

fix: Engagement Dashboard - values discrepancy with downloaded report - #31433

Merged
kodiakhq[bot] merged 5 commits into
developfrom
fix/engagement-dashboard-activeUsersSection
Jan 23, 2024
Merged

fix: Engagement Dashboard - values discrepancy with downloaded report#31433
kodiakhq[bot] merged 5 commits into
developfrom
fix/engagement-dashboard-activeUsersSection

Conversation

@hugocostadev

@hugocostadev hugocostadev commented Jan 12, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

The engagement dashboard is exhibiting inaccurate values for the latest day.

Using UTC timezone: The results exhibited in the graph are actually the results of the prior day (mismatch) (at least when compared to the report)

Using local timezone: MAU and WAU appears as 0; it must be at the minimum the same value of DAU

Issue(s)

Steps to test or reproduce

Further comments

WM-200

@hugocostadev hugocostadev added this to the 6.6 milestone Jan 12, 2024
@hugocostadev
hugocostadev requested a review from a team as a code owner January 12, 2024 16:04
@changeset-bot

changeset-bot Bot commented Jan 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2c3feb4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@hugocostadev hugocostadev changed the title fix: values discrepancy with downloaded report fix: Engagement Dashboard - values discrepancy with downloaded report Jan 12, 2024
@codecov

codecov Bot commented Jan 12, 2024

Copy link
Copy Markdown

Codecov Report

Attention: 2 lines in your changes are missing coverage. Please review.

Comparison is base (0c94ea3) 49.54% compared to head (2c3feb4) 49.57%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31433      +/-   ##
===========================================
+ Coverage    49.54%   49.57%   +0.03%     
===========================================
  Files         3318     3322       +4     
  Lines        81689    81741      +52     
  Branches     16795    16802       +7     
===========================================
+ Hits         40470    40527      +57     
+ Misses       36504    36500       -4     
+ Partials      4715     4714       -1     
Flag Coverage Δ
e2e 53.16% <66.66%> (+0.09%) ⬆️
e2e-api 40.73% <100.00%> (+0.01%) ⬆️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 23, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 23, 2024
@kodiakhq
kodiakhq Bot merged commit 3708609 into develop Jan 23, 2024
@kodiakhq
kodiakhq Bot deleted the fix/engagement-dashboard-activeUsersSection branch January 23, 2024 18:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants