Skip to content

fix: Editing room type or name doesn't always refresh the room - #31387

Merged
kodiakhq[bot] merged 6 commits into
developfrom
fix/roomInfoRedirect
Jan 5, 2024
Merged

fix: Editing room type or name doesn't always refresh the room#31387
kodiakhq[bot] merged 6 commits into
developfrom
fix/roomInfoRedirect

Conversation

@gabriellsh

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@gabriellsh
gabriellsh requested a review from a team as a code owner January 5, 2024 18:05
@changeset-bot

changeset-bot Bot commented Jan 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7ea14e2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 5, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (d5d7d50) 59.36% compared to head (7ea14e2) 49.39%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31387      +/-   ##
===========================================
- Coverage    59.36%   49.39%   -9.97%     
===========================================
  Files         1794     3312    +1518     
  Lines        34920    81366   +46446     
  Branches      7246    16657    +9411     
===========================================
+ Hits         20729    40188   +19459     
- Misses       12589    36477   +23888     
- Partials      1602     4701    +3099     
Flag Coverage Δ
e2e 52.60% <100.00%> (+0.07%) ⬆️
e2e-api 40.69% <ø> (?)
unit 76.67% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

yash-rajpal
yash-rajpal previously approved these changes Jan 5, 2024

@yash-rajpal yash-rajpal left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!!

yash-rajpal
yash-rajpal previously approved these changes Jan 5, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 5, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 5, 2024
@scuciatto scuciatto added this to the 6.6 milestone Jan 5, 2024
@kodiakhq
kodiakhq Bot merged commit 43335c8 into develop Jan 5, 2024
@kodiakhq
kodiakhq Bot deleted the fix/roomInfoRedirect branch January 5, 2024 20:58
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants