Skip to content

fix: LDAP Group filter doesn't work and throws "No Such Object" error on login - #31377

Merged
kodiakhq[bot] merged 8 commits into
developfrom
fix/ldap-group-filter
Jan 12, 2024
Merged

fix: LDAP Group filter doesn't work and throws "No Such Object" error on login#31377
kodiakhq[bot] merged 8 commits into
developfrom
fix/ldap-group-filter

Conversation

@matheusbsilva137

@matheusbsilva137 matheusbsilva137 commented Jan 4, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

  • Invalidate current binding after authenticating users (so that the bind process is performed again with the admin for the next action, such as searching for groups). Authenticated users may not have the read privileges needed to check if they are a member of a group, so that could cause the group membership check to fail.

Issue(s)

Fixes #15621
Fixes #15069

Steps to test or reproduce

Further comments

TC-1073

@changeset-bot

changeset-bot Bot commented Jan 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a47635c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 4, 2024

Copy link
Copy Markdown

Codecov Report

Attention: 4 lines in your changes are missing coverage. Please review.

Comparison is base (dd2967c) 49.60% compared to head (a47635c) 49.56%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31377      +/-   ##
===========================================
- Coverage    49.60%   49.56%   -0.04%     
===========================================
  Files         3308     3312       +4     
  Lines        81389    81460      +71     
  Branches     16697    16712      +15     
===========================================
+ Hits         40370    40379       +9     
- Misses       36328    36370      +42     
- Partials      4691     4711      +20     
Flag Coverage Δ
e2e 53.18% <ø> (-0.10%) ⬇️
e2e-api 40.67% <0.00%> (-0.01%) ⬇️
unit 76.51% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@matheusbsilva137 matheusbsilva137 added this to the 6.6 milestone Jan 4, 2024
@matheusbsilva137
matheusbsilva137 marked this pull request as ready for review January 5, 2024 12:52
@matheusbsilva137
matheusbsilva137 requested a review from a team as a code owner January 5, 2024 12:52

@pierre-lehnen-rc pierre-lehnen-rc left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Instead of creating a new connection, just validate the group before validating the password.

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 12, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 12, 2024
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Jan 12, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 12, 2024
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Jan 12, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 12, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 12, 2024
@kodiakhq
kodiakhq Bot merged commit afd5fdd into develop Jan 12, 2024
@kodiakhq
kodiakhq Bot deleted the fix/ldap-group-filter branch January 12, 2024 21:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Login NoSuchObjectError when LDAP_Sync_User_Data_Groups set to true Can't enable LDAP Group Validation

3 participants