Skip to content

feat: Implemented InlineCode handling in Bold, Italic and Strike - #31349

Merged
ggazzo merged 10 commits into
RocketChat:developfrom
sdey09:ISSUE-1213
Jan 24, 2024
Merged

feat: Implemented InlineCode handling in Bold, Italic and Strike#31349
ggazzo merged 10 commits into
RocketChat:developfrom
sdey09:ISSUE-1213

Conversation

@sdey09

@sdey09 sdey09 commented Jan 1, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

The required result for the corresponding codes changes.
Screenshot from 2024-01-01 17-37-53

Issue(s)

Fixes #30185

Steps to test or reproduce

Please follow the #30185

https://rocketchat.atlassian.net/browse/SUP-285

@changeset-bot

changeset-bot Bot commented Jan 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ffce14b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-services Patch
@rocket.chat/core-typings Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/rest-typings Patch
@rocket.chat/pdf-worker Patch
rocketchat-services Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/presence Patch
@rocket.chat/cron Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/ddp-client Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@hugocostadev
hugocostadev self-requested a review January 8, 2024 19:36
@codecov

codecov Bot commented Jan 8, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (265faec) 49.55% compared to head (ffce14b) 45.37%.
Report is 1 commits behind head on develop.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31349      +/-   ##
===========================================
- Coverage    49.55%   45.37%   -4.19%     
===========================================
  Files         3325     3218     -107     
  Lines        81733    75576    -6157     
  Branches     16861    15486    -1375     
===========================================
- Hits         40502    34289    -6213     
- Misses       36496    36828     +332     
+ Partials      4735     4459     -276     
Flag Coverage Δ
e2e 53.27% <ø> (-0.02%) ⬇️
e2e-api 40.62% <ø> (-0.01%) ⬇️
unit ?

Flags with carried forward coverage won't be shown. Click here to find out more.

@hugocostadev hugocostadev left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM 🚀 thanks for your contribution

We need to wait the RocketChat/fuselage#1256 PR be merged and than update the message-parser package to actually fix the issue

image

@hugocostadev

Copy link
Copy Markdown
Contributor

Hey man, it was merged! Thanks a lot!
If possible, update the package version when it gets updated in the npm, pelase.

@rocket.chat/message-parser - npm

@sdey09

sdey09 commented Jan 18, 2024

Copy link
Copy Markdown
Contributor Author

Hi @hugocostadev , thanks a lot for the changes. I have been checking the version might have missed it in the couple of days. Let me know if any changes are required from my side.

@sdey09
sdey09 requested review from a team as code owners January 18, 2024 11:52
@hugocostadev hugocostadev added this to the 6.6 milestone Jan 19, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 19, 2024
hugocostadev
hugocostadev previously approved these changes Jan 19, 2024
tapiarafael
tapiarafael previously approved these changes Jan 23, 2024
@ggazzo ggazzo changed the title fix: Implemented InlineCode handling in Bold, Italic and Strike feat: Implemented InlineCode handling in Bold, Italic and Strike Jan 24, 2024
@ggazzo
ggazzo dismissed stale reviews from tapiarafael and hugocostadev via 047eb98 January 24, 2024 15:35
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 24, 2024
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Jan 24, 2024
@ggazzo
ggazzo merged commit dbb08ef into RocketChat:develop Jan 24, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Development

Successfully merging this pull request may close these issues.

Markdown: strikethrough text does not work with the code span

4 participants