fix: Displaying wrong composer for archived room - #31292
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 7abc899 The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAttention:
Additional details and impacted files@@ Coverage Diff @@
## develop #31292 +/- ##
============================================
- Coverage 59.37% 49.31% -10.07%
============================================
Files 1782 3301 +1519
Lines 34720 81131 +46411
Branches 7208 16593 +9385
============================================
+ Hits 20616 40010 +19394
- Misses 12529 36462 +23933
- Partials 1575 4659 +3084
Flags with carried forward coverage won't be shown. Click here to find out more. |
dougfabris
marked this pull request as ready for review
December 21, 2023 14:37
hugocostadev
approved these changes
Dec 21, 2023
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
When the public room has
joinCodeand it is archived, it's displaying the wrong composer, it should show the archived composer insteadOld behavior
New behavior
Issue(s)
Steps to test or reproduce
Further comments
TC-853