Skip to content

fix: Avoid creating a visitor when token is an empty string - #31287

Merged
kodiakhq[bot] merged 3 commits into
developfrom
fix/emtpy-token
Dec 22, 2023
Merged

fix: Avoid creating a visitor when token is an empty string#31287
kodiakhq[bot] merged 3 commits into
developfrom
fix/emtpy-token

Conversation

@KevLehman

@KevLehman KevLehman commented Dec 20, 2023

Copy link
Copy Markdown
Member

SUP-309

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Dec 20, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f8acdd8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 20, 2023

Copy link
Copy Markdown

Codecov Report

Attention: 1 lines in your changes are missing coverage. Please review.

Comparison is base (d70abcc) 49.31% compared to head (f8acdd8) 47.44%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31287      +/-   ##
===========================================
- Coverage    49.31%   47.44%   -1.87%     
===========================================
  Files         3299     3095     -204     
  Lines        81120    77741    -3379     
  Branches     16593    15855     -738     
===========================================
- Hits         40006    36886    -3120     
- Misses       36456    36459       +3     
+ Partials      4658     4396     -262     
Flag Coverage Δ
e2e 46.84% <ø> (-5.76%) ⬇️
e2e-api 40.63% <75.00%> (+<0.01%) ⬆️
unit 76.39% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Dec 22, 2023
@KevLehman
KevLehman marked this pull request as ready for review December 22, 2023 13:54
@KevLehman
KevLehman requested review from a team as code owners December 22, 2023 13:54
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 22, 2023
@kodiakhq
kodiakhq Bot merged commit b6b7198 into develop Dec 22, 2023
@kodiakhq
kodiakhq Bot deleted the fix/emtpy-token branch December 22, 2023 14:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants