Skip to content

fix: file type filter not working on room's file list - #31228

Merged
ggazzo merged 1 commit into
developfrom
fix/channel-file-list-type-filter
Dec 13, 2023
Merged

fix: file type filter not working on room's file list#31228
ggazzo merged 1 commit into
developfrom
fix/channel-file-list-type-filter

Conversation

@pierre-lehnen-rc

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Dec 13, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 11de41b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 13, 2023

Copy link
Copy Markdown

Codecov Report

Merging #31228 (11de41b) into develop (1bcb57c) will decrease coverage by 28.49%.
Report is 4 commits behind head on develop.
The diff coverage is n/a.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #31228       +/-   ##
============================================
- Coverage    77.38%   48.90%   -28.49%     
============================================
  Files          276     3270     +2994     
  Lines         8715    80305    +71590     
  Branches      1607    16389    +14782     
============================================
+ Hits          6744    39270    +32526     
- Misses        1565    36384    +34819     
- Partials       406     4651     +4245     
Flag Coverage Δ
e2e 51.39% <ø> (∅)
e2e-api 40.69% <ø> (?)
unit 77.36% <ø> (-0.02%) ⬇️

Flags with carried forward coverage won't be shown. Click here to find out more.

@pierre-lehnen-rc
pierre-lehnen-rc marked this pull request as ready for review December 13, 2023 17:58
@pierre-lehnen-rc
pierre-lehnen-rc requested a review from a team as a code owner December 13, 2023 17:58
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Dec 13, 2023
@dougfabris dougfabris added this to the 6.6 milestone Dec 13, 2023

@dougfabris dougfabris left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@ggazzo
ggazzo merged commit 097f64b into develop Dec 13, 2023
@ggazzo
ggazzo deleted the fix/channel-file-list-type-filter branch December 13, 2023 19:14
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants