Skip to content

fix: connected users search on spotlight for user mentions - #31134

Merged
kodiakhq[bot] merged 10 commits into
developfrom
fix/spotlight-outside-users
Dec 15, 2023
Merged

fix: connected users search on spotlight for user mentions#31134
kodiakhq[bot] merged 10 commits into
developfrom
fix/spotlight-outside-users

Conversation

@yash-rajpal

@yash-rajpal yash-rajpal commented Dec 1, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

TC-1050
TC-1052

@yash-rajpal
yash-rajpal requested a review from a team as a code owner December 1, 2023 13:08
@changeset-bot

changeset-bot Bot commented Dec 1, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1190a03

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 1, 2023

Copy link
Copy Markdown

Codecov Report

Merging #31134 (1190a03) into develop (13ef78a) will decrease coverage by 28.07%.
The diff coverage is 50.00%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #31134       +/-   ##
============================================
- Coverage    77.40%   49.33%   -28.07%     
============================================
  Files          277     3297     +3020     
  Lines         8731    80908    +72177     
  Branches      1620    16508    +14888     
============================================
+ Hits          6758    39915    +33157     
- Misses        1565    36339    +34774     
- Partials       408     4654     +4246     
Flag Coverage Δ
e2e 52.65% <ø> (∅)
e2e-api 40.68% <50.00%> (?)
unit 77.40% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@scuciatto scuciatto added this to the 6.6 milestone Dec 5, 2023
Comment thread apps/meteor/server/lib/spotlight.js Outdated
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Dec 15, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 15, 2023
@kodiakhq
kodiakhq Bot merged commit 83bcf04 into develop Dec 15, 2023
@kodiakhq
kodiakhq Bot deleted the fix/spotlight-outside-users branch December 15, 2023 15:34
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants