fix: Discussion messages deleted despite the "Do not delete discussion messages" retention policy enabled - #31113
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 84655ea The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## develop #31113 +/- ##
===========================================
+ Coverage 57.18% 58.25% +1.07%
===========================================
Files 880 1760 +880
Lines 19925 33639 +13714
Branches 3719 6907 +3188
===========================================
+ Hits 11395 19598 +8203
- Misses 7841 12500 +4659
- Partials 689 1541 +852
Flags with carried forward coverage won't be shown. Click here to find out more. |
matheusbsilva137
approved these changes
Dec 1, 2023
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Discussion messages were being deleted despite the "Do not delete discussion messages" retention policy being enabled.
Altered the message deletion query to accurately respect the “Do not delete discussion messages” setting by excluding discussion rooms from the deletion process.
Issue(s)
Steps to test or reproduce
Prerequisite:
Steps to reproduce:
Expected Behavior:
The discussion messages should be kept
Actual Behavior:
If the user enables the setting “Do not prune discussion messages” it’s expected to be applied for all the discussion messages, not only for the creation one.
Further comments
TC-805