fix: mention channel redirecting to own DM - #31099
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: a26512d The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## develop #31099 +/- ##
===========================================
- Coverage 56.72% 55.29% -1.43%
===========================================
Files 883 1585 +702
Lines 19990 31120 +11130
Branches 3689 6309 +2620
===========================================
+ Hits 11340 17209 +5869
- Misses 7956 12565 +4609
- Partials 694 1346 +652
Flags with carried forward coverage won't be shown. Click here to find out more. |
yash-rajpal
marked this pull request as ready for review
November 29, 2023 13:32
yash-rajpal
approved these changes
Nov 29, 2023
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments
https://rocketchat.atlassian.net/browse/TC-1017