fix: UnhandledPromiseRejection on PUT livechat/departments/:_id endpoint - #31049
Merged
Conversation
🦋 Changeset detectedLatest commit: 3ab8dda The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## develop #31049 +/- ##
===========================================
+ Coverage 46.32% 51.18% +4.86%
===========================================
Files 766 807 +41
Lines 14676 15139 +463
Branches 2753 2819 +66
===========================================
+ Hits 6798 7749 +951
+ Misses 7479 6957 -522
- Partials 399 433 +34
Flags with carried forward coverage won't be shown. Click here to find out more. |
aleksandernsilva
previously approved these changes
Nov 22, 2023
aleksandernsilva
previously approved these changes
Nov 22, 2023
MarcosSpessatto
previously approved these changes
Nov 22, 2023
KevLehman
marked this pull request as ready for review
November 22, 2023 18:00
KevLehman
dismissed stale reviews from MarcosSpessatto and aleksandernsilva
via
November 22, 2023 20:22
6aa77c2
aleksandernsilva
approved these changes
Nov 22, 2023
MarcosSpessatto
approved these changes
Nov 22, 2023
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments