Skip to content

fix: Message mentions displayed in the UI are outdated after changing username - #30625

Closed
matheusbsilva137 wants to merge 2 commits into
developfrom
fix/change-username-doesnt-update-mentions-ui
Closed

fix: Message mentions displayed in the UI are outdated after changing username#30625
matheusbsilva137 wants to merge 2 commits into
developfrom
fix/change-username-doesnt-update-mentions-ui

Conversation

@matheusbsilva137

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

  • Unset markdown (md) field when updating messages that mention the username changed (so that the client tries to parse it again with the updated content).

Issue(s)

Steps to test or reproduce

Further comments

SUP-378

@changeset-bot

changeset-bot Bot commented Oct 11, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 815a146

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Oct 11, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30625 (815a146) into develop (0cdf97d) will decrease coverage by 0.05%.
Report is 1 commits behind head on develop.
The diff coverage is n/a.

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #30625      +/-   ##
===========================================
- Coverage    51.28%   51.24%   -0.05%     
===========================================
  Files          811      809       -2     
  Lines        15078    15140      +62     
  Branches      2754     2775      +21     
===========================================
+ Hits          7733     7758      +25     
- Misses        6935     6954      +19     
- Partials       410      428      +18     
Flag Coverage Δ
e2e 48.57% <ø> (+0.01%) ⬆️
unit 64.80% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@yash-rajpal yash-rajpal left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Shouldn't we parse the newMessage and update the md property?

@matheusbsilva137

Copy link
Copy Markdown
Contributor Author

Handled in #31860

@tassoevan
tassoevan deleted the fix/change-username-doesnt-update-mentions-ui branch August 1, 2025 15:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants