fix: case-sensitive json marshalling and unmarshalling - #1863
Merged
Merged
Conversation
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1863 +/- ##
=======================================
Coverage 84.38% 84.38%
=======================================
Files 130 130
Lines 18114 18114
=======================================
Hits 15285 15285
Misses 1883 1883
Partials 946 946 ☔ View full report in Codecov by Sentry. 🚀 New features to boost your workflow:
|
Contributor
|
Can we make this configurable? it may break requests that are passing unknown elements or fields not defined on struct |
Member
Author
|
I've updated the change to focus on case-sensitivity, which shouldn't affect unknown fields, etc, that we rely on. |
Signed-off-by: Ignasi Barrera <nacx@apache.org>
Signed-off-by: Ignasi Barrera <nacx@apache.org>
Signed-off-by: Ignasi Barrera <nacx@apache.org>
Member
Author
|
/retest |
Signed-off-by: Ignasi Barrera <nacx@apache.org>
nacx
marked this pull request as ready for review
April 12, 2026 11:21
Member
Author
|
@yuzisun can you take a look and approve? This is passing all tests and fixes the security advisory on MCP. |
mathetake
reviewed
Apr 13, 2026
mathetake
left a comment
Contributor
There was a problem hiding this comment.
needs benchmark comparison just in case
Member
Author
|
Contributor
|
Nice |
Member
Author
|
@mathetake can you stamp? :) |
mathetake
approved these changes
Apr 15, 2026
nacx
enabled auto-merge (squash)
April 15, 2026 20:15
Member
Author
|
/retest |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description
Use strict json marshalling and unmarshalling configuration.
Related Issues/PRs (if applicable)
N/A
Special notes for reviewers (if applicable)
N/A