Update golang.org/x/exp digest to 746e56f - #74
red-hat-konflux[bot] wants to merge 1 commit into
Conversation
ℹ Artifact update noticeFile name: v2/cmd/asoctl/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
File name: v2/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
File name: v2/tools/generator/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
|
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
54c9ee8 to
be8e0be
Compare
be8e0be to
a67ab1f
Compare
a67ab1f to
ce46e05
Compare
ce46e05 to
23aa68b
Compare
ℹ️ Artifact update noticeFile name: v2/cmd/asoctl/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
File name: v2/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
File name: v2/tools/generator/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
|
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
23aa68b to
8fca0ef
Compare
Renovate Ignore NotificationBecause you closed this PR without merging, Renovate will ignore this update. You will not get PRs for the If you accidentally closed this PR, or if you changed your mind: rename this PR to get a fresh replacement PR. |
This PR contains the following updates:
054e65f→746e56fWarning
Some dependencies could not be looked up. Check the warning logs for more information.
Configuration
📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.