Skip to content

[UnifiedTree]: Add nightly hicache ci for dsa model - #25348

Merged
ispobock merged 2 commits into
sgl-project:mainfrom
hzh0425:hybrid_tree/hicache_ci_glm5
May 15, 2026
Merged

ispobock merged 2 commits into
sgl-project:mainfrom
hzh0425:hybrid_tree/hicache_ci_glm5

Conversation

@hzh0425

@hzh0425 hzh0425 commented May 15, 2026

Copy link
Copy Markdown
Collaborator

Motivation

Modifications

Accuracy Tests

Speed Tests and Profiling

Checklist

Review and Merge Process

  1. Ping Merge Oncalls to start the process. See the PR Merge Process.
  2. Get approvals from CODEOWNERS and other reviewers.
  3. Trigger CI tests with comments or contact authorized users to do so.
    • Common commands include /tag-and-rerun-ci, /tag-run-ci-label, /rerun-failed-ci
  4. After green CI and required approvals, ask Merge Oncalls or people with Write permission to merge the PR.

@gemini-code-assist

Copy link
Copy Markdown
Contributor

Warning

You have reached your daily quota limit. Please wait up to 24 hours and I will start processing your requests again!

@github-actions github-actions Bot added the hicache Hierarchical Caching for SGLang label May 15, 2026
@hzh0425

hzh0425 commented May 15, 2026

Copy link
Copy Markdown
Collaborator Author

/rerun-test test/registered/radix_cache/test_unified_radix_hicache_kl.py

@github-actions

github-actions Bot commented May 15, 2026

Copy link
Copy Markdown
Contributor

🚀 8-gpu-h200 (1 test): ✅ View workflow run

cd test/ && python3 registered/radix_cache/test_unified_radix_hicache_kl.py

@hzh0425

hzh0425 commented May 15, 2026

Copy link
Copy Markdown
Collaborator Author

/rerun-test test/registered/radix_cache/test_unified_radix_hicache_kl.py

@github-actions

github-actions Bot commented May 15, 2026

Copy link
Copy Markdown
Contributor

🚀 8-gpu-h200 (1 test): ✅ View workflow run

cd test/ && python3 registered/radix_cache/test_unified_radix_hicache_kl.py

@hzh0425

hzh0425 commented May 15, 2026

Copy link
Copy Markdown
Collaborator Author
image

@ispobock
ispobock merged commit fe7a5dd into sgl-project:main May 15, 2026
77 of 86 checks passed
Chronostasys pushed a commit to MindLab-Research/sglang that referenced this pull request Aug 24, 2026
jakki-amd pushed a commit to jakki-amd/sglang that referenced this pull request Sep 9, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

hicache Hierarchical Caching for SGLang

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants