Skip to content

[sglang-miles] chore: add bias for base layer with lora - #22846

Merged
yushengsu-thu merged 1 commit into
sgl-project:sglang-milesfrom
gongyisheng:sglang-miles-update-weight-refactory2
Apr 16, 2026
Merged

yushengsu-thu merged 1 commit into
sgl-project:sglang-milesfrom
gongyisheng:sglang-miles-update-weight-refactory2

Conversation

@gongyisheng

@gongyisheng gongyisheng commented Apr 15, 2026

Copy link
Copy Markdown
Contributor

Motivation

ref to main PR: #22169

Modifications

Accuracy Tests

Speed Tests and Profiling

Checklist

Review and Merge Process

  1. Ping Merge Oncalls to start the process. See the PR Merge Process.
  2. Get approvals from CODEOWNERS and other reviewers.
  3. Trigger CI tests with comments or contact authorized users to do so.
    • Common commands include /tag-and-rerun-ci, /tag-run-ci-label, /rerun-failed-ci
  4. After green CI and required approvals, ask Merge Oncalls or people with Write permission to merge the PR.

@gemini-code-assist

Copy link
Copy Markdown
Contributor

Warning

You have reached your daily quota limit. Please wait up to 24 hours and I will start processing your requests again!

@gongyisheng gongyisheng changed the title chore: add bias for base layer with lora [sglang-miles] chore: add bias for base layer with lora Apr 15, 2026
@yushengsu-thu
yushengsu-thu merged commit 3c43f43 into sgl-project:sglang-miles Apr 16, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants