Skip to content

[sglang-miles] tmp fix for vlm training: use legacy_load_mm_data for qwen-vl - #18781

Merged
yueming-yuan merged 1 commit into
sglang-milesfrom
fix260212_vl
Feb 13, 2026
Merged

yueming-yuan merged 1 commit into
sglang-milesfrom
fix260212_vl

Conversation

@yueming-yuan

Copy link
Copy Markdown
Collaborator

Motivation

Modifications

Accuracy Tests

Benchmarking and Profiling

Checklist

Review Process

  1. Ping Merge Oncalls to start the PR flow. See the PR Merge Process.
  2. Get approvals from CODEOWNERS and other reviewers.
  3. Trigger CI tests with comments or contact authorized users to do so.
    • /tag-run-ci-label, /rerun-failed-ci, /tag-and-rerun-ci
  4. After green CI and required approvals, ask Merge Oncalls to merge.

@gemini-code-assist

Copy link
Copy Markdown
Contributor

Warning

You have reached your daily quota limit. Please wait up to 24 hours and I will start processing your requests again!

@yueming-yuan
yueming-yuan merged commit e326729 into sglang-miles Feb 13, 2026
1 check passed
@yueming-yuan
yueming-yuan deleted the fix260212_vl branch February 13, 2026 06:31
zianglih pushed a commit to zianglih/sglang that referenced this pull request Feb 28, 2026
yueming-yuan added a commit that referenced this pull request Mar 5, 2026
yueming-yuan added a commit that referenced this pull request Mar 27, 2026
yueming-yuan added a commit that referenced this pull request Apr 6, 2026
JessicaJiang-123 pushed a commit to JessicaJiang-123/sglang that referenced this pull request Jun 21, 2026
nanjiangwill pushed a commit to nanjiangwill/sglang that referenced this pull request Jul 7, 2026
XinyuJiangCMU pushed a commit to XinyuJiangCMU/sglang that referenced this pull request Jul 17, 2026
Arist12 pushed a commit to Arist12/sglang that referenced this pull request Aug 27, 2026
Kh4L pushed a commit to Kh4L/sglang that referenced this pull request Sep 8, 2026
Zhichenzzz pushed a commit that referenced this pull request Sep 19, 2026
(cherry picked from commit c4b6f74)
(cherry picked from commit 049efd7)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant