Skip to content

Update cargo submodule - #161492

Merged
rust-bors[bot] merged 1 commit into
rust-lang:mainfrom
weihanglo:update-cargo
Aug 22, 2026
Merged

Update cargo submodule#161492
rust-bors[bot] merged 1 commit into
rust-lang:mainfrom
weihanglo:update-cargo

Conversation

@weihanglo

Copy link
Copy Markdown
Member

5 commits in 514c56dd7321eecbfdcf9b6479519cf4edfab906..e8cb624d5701824f46a2ec5873cfd59ee3d2f66c
2026-08-19 00:31:04 +0000 to 2026-08-22 00:23:45 +0000

r? ghost

@rustbot rustbot added the S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. label Aug 22, 2026
@rustbot

rustbot commented Aug 22, 2026

Copy link
Copy Markdown
Collaborator

⚠️ Warning ⚠️

@weihanglo

Copy link
Copy Markdown
Member Author

@bors r+ rollup=never p=1

@rust-bors

rust-bors Bot commented Aug 22, 2026

Copy link
Copy Markdown
Contributor

📌 Commit 0c61f07 has been approved by weihanglo

It is now in the queue for this repository.

@rust-bors rust-bors Bot added S-waiting-on-bors Status: Waiting on bors to run and complete tests. Bors will change the label on completion. and removed S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. labels Aug 22, 2026
@JonathanBrouwer

Copy link
Copy Markdown
Member

@bors p=6 scheduling

@rust-bors

This comment has been minimized.

@rust-bors rust-bors Bot added merged-by-bors This PR was explicitly merged by bors. and removed S-waiting-on-bors Status: Waiting on bors to run and complete tests. Bors will change the label on completion. labels Aug 22, 2026
@rust-bors

rust-bors Bot commented Aug 22, 2026

Copy link
Copy Markdown
Contributor

☀️ Test successful - CI
Approved by: weihanglo
Duration: 2h 57m 4s
Pushing 8536f4d to main...

@rust-bors
rust-bors Bot merged commit 8536f4d into rust-lang:main Aug 22, 2026
14 checks passed
@rustbot rustbot added this to the 1.100.0 milestone Aug 22, 2026
@github-actions

Copy link
Copy Markdown
Contributor
What is this? This is an experimental post-merge analysis report that shows differences in test outcomes between the merged PR and its parent PR.

Comparing 3009fdd (parent) -> 8536f4d (this PR)

Test differences

Show 81 test diffs

Stage 2

  • compiler::fingerprint::dirty_reason::json_schema::checksum_use_changed: pass -> [missing] (J0)
  • compiler::fingerprint::dirty_reason::json_schema::fingerprint_method_changed: [missing] -> pass (J0)
  • freshness::bust_patched_dep: pass -> [missing] (J0)
  • freshness::checksum_build_compatible_with_mtime_build: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness::dirty_both_lib_and_test: pass -> [missing] (J0)
  • freshness::fingerprint_cleaner_does_not_rebuild: pass -> [missing] (J0)
  • freshness::incremental_build_script_execution_got_new_mtime_and_cargo_check: pass -> [missing] (J0)
  • freshness::modify_only_some_files: pass -> [missing] (J0)
  • freshness::modifying_and_moving: pass -> [missing] (J0)
  • freshness::move_target_directory_with_path_deps: pass -> [missing] (J0)
  • freshness::no_rebuild_if_build_artifacts_move_backwards_in_time: pass -> [missing] (J0)
  • freshness::no_rebuild_when_rename_dir: pass -> [missing] (J0)
  • freshness::rebuild_if_build_artifacts_move_forward_in_time: pass -> [missing] (J0)
  • freshness::rebuild_on_mid_build_file_modification: pass -> [missing] (J0)
  • freshness::rebuild_sub_package_then_while_package: pass -> [missing] (J0)
  • freshness::rebuild_tests_if_lib_changes: pass -> [missing] (J0)
  • freshness::rename_with_path_deps: pass -> [missing] (J0)
  • freshness::script_fails_stay_dirty: pass -> [missing] (J0)
  • freshness::simulated_docker_deps_stay_cached: pass -> [missing] (J0)
  • freshness::skip_mtime_check_in_selected_cargo_home_subdirs: pass -> [missing] (J0)
  • freshness::symlink_to_package: pass -> [missing] (J0)
  • freshness::update_dependency_mtime_does_not_rebuild: pass -> [missing] (J0)
  • freshness::use_mtime_cache_in_cargo_home: pass -> [missing] (J0)
  • freshness::verify_source_before_recompile: pass -> [missing] (J0)
  • freshness_checksum::change_panic_mode: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_bin_features_caches_targets: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_bin_paths_common_target_features_caches_targets: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_lib_features_caches_targets: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_linker: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_profiles_caches_targets: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_rustc_extra_flags_is_cached: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::changing_rustflags_is_cached: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::channel_shares_filenames: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::checksum_build_compatible_with_mtime_build: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::dont_rebuild_based_on_plugins: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::edition_change_invalidates: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::env_build_script_no_rebuild: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::env_in_code_causes_rebuild: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::fingerprint_cleaner_does_not_rebuild: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::linking_interrupted: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::modify_only_some_files: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::no_rebuild_if_build_artifacts_move_forward_in_time: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::no_rebuild_transitive_target_deps: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::path_dev_dep_registry_updates: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::rebuild_if_environment_changes: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::rerun_if_changed_in_dep: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::rerun_if_changes: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::reuse_panic_build_dep_test: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::reuse_panic_pm: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::reuse_shared_build_dep: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::reuse_workspace_lib: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::same_build_dir_cached_packages: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::simulated_docker_deps_stay_cached: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::symlink_to_package: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::unused_optional_dep: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J0)
  • freshness_checksum::update_dependency_mtime_does_not_rebuild: [missing] -> ignore (requires -Zchecksum-hash-algorithm) (J0)
  • freshness_checksum::warn_on_missing_feature: [missing] -> pass (J0)
  • freshness_mtime::bust_patched_dep: [missing] -> pass (J0)
  • freshness_mtime::dirty_both_lib_and_test: [missing] -> pass (J0)
  • freshness_mtime::fingerprint_cleaner_does_not_rebuild: [missing] -> pass (J0)
  • freshness_mtime::incremental_build_script_execution_got_new_mtime_and_cargo_check: [missing] -> pass (J0)
  • freshness_mtime::modify_only_some_files: [missing] -> pass (J0)
  • freshness_mtime::modifying_and_moving: [missing] -> pass (J0)
  • freshness_mtime::move_target_directory_with_path_deps: [missing] -> pass (J0)
  • freshness_mtime::no_rebuild_if_build_artifacts_move_backwards_in_time: [missing] -> pass (J0)
  • freshness_mtime::no_rebuild_when_rename_dir: [missing] -> pass (J0)
  • freshness_mtime::rebuild_if_build_artifacts_move_forward_in_time: [missing] -> pass (J0)
  • freshness_mtime::rebuild_on_mid_build_file_modification: [missing] -> pass (J0)
  • freshness_mtime::rebuild_sub_package_then_while_package: [missing] -> pass (J0)
  • freshness_mtime::rebuild_tests_if_lib_changes: [missing] -> pass (J0)
  • freshness_mtime::rename_with_path_deps: [missing] -> pass (J0)
  • freshness_mtime::script_fails_stay_dirty: [missing] -> pass (J0)
  • freshness_mtime::simulated_docker_deps_stay_cached: [missing] -> pass (J0)
  • freshness_mtime::skip_mtime_check_in_selected_cargo_home_subdirs: [missing] -> pass (J0)
  • freshness_mtime::symlink_to_package: [missing] -> pass (J0)
  • freshness_mtime::update_dependency_mtime_does_not_rebuild: [missing] -> pass (J0)
  • freshness_mtime::use_mtime_cache_in_cargo_home: [missing] -> pass (J0)
  • freshness_mtime::verify_source_before_recompile: [missing] -> pass (J0)
  • freshness_checksum::lld_is_fresh: ignore (requires -Zchecksum-hash-algorithm) -> [missing] (J1)

Additionally, 2 doctest diffs were found. These are ignored, as they are noisy.

Job group index

Test dashboard

Run

cargo run --manifest-path src/ci/citool/Cargo.toml -- \
    test-dashboard 8536f4de90e5e781c06e584f8a88de860adc798b --output-dir test-dashboard

And then open test-dashboard/index.html in your browser to see an overview of all executed tests.

Job duration changes

  1. x86_64-gnu-stdlib-semver-check: 18m 16s -> 10m 22s (-43.2%)
  2. optional-x86_64-gnu-autodiff: 57m 10s -> 33m 37s (-41.2%)
  3. dist-x86_64-llvm-mingw: 1h 35m -> 2h 12m (+39.0%)
  4. dist-apple-various: 1h 40m -> 2h 18m (+38.4%)
  5. i686-msvc-1: 3h 1m -> 2h 3m (-32.3%)
  6. x86_64-gnu-stable: 2h 33m -> 1h 44m (-31.8%)
  7. x86_64-gnu: 2h 40m -> 1h 49m (-31.8%)
  8. x86_64-gnu-llvm-21-1: 1h 1m -> 42m 26s (-30.8%)
  9. i686-gnu-nopt-2: 1h 49m -> 2h 21m (+29.4%)
  10. pr-check-2: 43m 22s -> 31m 55s (-26.4%)
How to interpret the job duration changes?

Job durations can vary a lot, based on the actual runner instance
that executed the job, system noise, invalidated caches, etc. The table above is provided
mostly for t-infra members, for simpler debugging of potential CI slow-downs.

@rust-timer

Copy link
Copy Markdown
Collaborator

Finished benchmarking commit (8536f4d): comparison URL.

Overall result: no relevant changes - no action needed

@rustbot label: -perf-regression

Instruction count

This perf run didn't have relevant results for this metric.

Max RSS (memory usage)

Results (primary 3.8%)

A less reliable metric. May be of interest, but not used to determine the overall result above.

mean range count
Regressions ❌
(primary)
3.8% [3.8%, 3.8%] 1
Regressions ❌
(secondary)
- - 0
Improvements ✅
(primary)
- - 0
Improvements ✅
(secondary)
- - 0
All ❌✅ (primary) 3.8% [3.8%, 3.8%] 1

Cycles

Results (primary -2.6%, secondary -2.3%)

A less reliable metric. May be of interest, but not used to determine the overall result above.

mean range count
Regressions ❌
(primary)
2.4% [2.4%, 2.4%] 1
Regressions ❌
(secondary)
- - 0
Improvements ✅
(primary)
-7.5% [-7.5%, -7.5%] 1
Improvements ✅
(secondary)
-2.3% [-2.6%, -2.0%] 3
All ❌✅ (primary) -2.6% [-7.5%, 2.4%] 2

Binary size

This perf run didn't have relevant results for this metric.

Bootstrap: 468.178s -> 468.334s (0.03%)
Artifact size: 400.13 MiB -> 400.18 MiB (0.01%)

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

merged-by-bors This PR was explicitly merged by bors.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants