Skip to content
This repository was archived by the owner on Sep 9, 2026. It is now read-only.

Bump osac-fulfillment-service: NO-ISSUE: Add builtin system and shared organizations - #191

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
bump/osac-fulfillment-service/pr-596
May 28, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
bump/osac-fulfillment-service/pr-596

Conversation

@osac-dev-bot

Copy link
Copy Markdown
Contributor

Bumps osac-fulfillment-service submodule.

Original PR: osac-project/fulfillment-service#596
Author: @jhernand
Merge commit: c23f884678ca6e0dd2b41b4309dd0e082bd4f31f

@openshift-ci-robot

Copy link
Copy Markdown

@osac-dev-bot: This pull request explicitly references no jira issue.

Details

In response to this:

Bumps osac-fulfillment-service submodule.

Original PR: osac-project/fulfillment-service#596
Author: @jhernand
Merge commit: c23f884678ca6e0dd2b41b4309dd0e082bd4f31f

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from akshaynadkarni and jhernand May 28, 2026 05:35
@openshift-ci openshift-ci Bot added the lgtm label May 28, 2026
@openshift-ci

openshift-ci Bot commented May 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jhernand, osac-dev-bot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 227621c into main May 28, 2026
9 of 10 checks passed
@openshift-ci

openshift-ci Bot commented May 28, 2026

Copy link
Copy Markdown

@osac-dev-bot: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-vmaas d2a6951 link unknown /test e2e-vmaas

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants