Skip to content

chore(release): switch image-reference tag to operator-lifecycle-manager - #612

Merged
openshift-merge-robot merged 1 commit into
operator-framework:masterfrom
ecordell:image-refs
Dec 10, 2018
Merged

chore(release): switch image-reference tag to operator-lifecycle-manager#612
openshift-merge-robot merged 1 commit into
operator-framework:masterfrom
ecordell:image-refs

Conversation

@ecordell

Copy link
Copy Markdown
Member

The OCP image name is tied to the image stream tag, so swapping it to what ocp expects

@openshift-ci-robot openshift-ci-robot added the size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. label Dec 10, 2018
@ecordell

Copy link
Copy Markdown
Member Author

/approve

@openshift-ci-robot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ecordell

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 10, 2018
@njhale

njhale commented Dec 10, 2018

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Dec 10, 2018
@openshift-merge-robot
openshift-merge-robot merged commit af566ee into operator-framework:master Dec 10, 2018
@ecordell
ecordell deleted the image-refs branch December 10, 2018 20:41
ecordell pushed a commit to ecordell/operator-lifecycle-manager that referenced this pull request Mar 8, 2019
chore(release): switch image-reference tag to operator-lifecycle-manager
@njhale njhale added the kind/feature Categorizes issue or PR as related to a new feature. label Mar 19, 2019
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/feature Categorizes issue or PR as related to a new feature. lgtm Indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants