Skip to content

OCPBUGS-24061: Keep CSI operators progressing=true during DaemonSet rollout - #230

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
dobsonj:OCPBUGS-24061
May 10, 2024
Merged

OCPBUGS-24061: Keep CSI operators progressing=true during DaemonSet rollout#230
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
dobsonj:OCPBUGS-24061

Conversation

@dobsonj

@dobsonj dobsonj commented May 7, 2024

Copy link
Copy Markdown
Member

https://issues.redhat.com/browse/OCPBUGS-24061

Bump library-go to get openshift/library-go#1734

go get github.com/openshift/library-go && go mod tidy && go mod vendor

/cc @openshift/storage

@openshift-ci
openshift-ci Bot requested a review from a team May 7, 2024 22:22
@openshift-ci-robot openshift-ci-robot added the jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. label May 7, 2024
@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label May 7, 2024
@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels May 7, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dobsonj: This pull request references Jira Issue OCPBUGS-24061, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

No GitHub users were found matching the public email listed for the QA contact in Jira (wduan@redhat.com), skipping review request.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

https://issues.redhat.com/browse/OCPBUGS-24061

Bump library-go to get openshift/library-go#1734

/cc @openshift/storage

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented May 7, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dobsonj

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 7, 2024
@dobsonj

dobsonj commented May 7, 2024

Copy link
Copy Markdown
Member Author

/payload-job periodic-ci-openshift-release-master-nightly-4.16-e2e-vsphere-ovn-techpreview-serial

@openshift-ci

openshift-ci Bot commented May 7, 2024

Copy link
Copy Markdown
Contributor

@dobsonj: trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-vsphere-ovn-techpreview-serial

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/af2abd90-0cc1-11ef-9fcd-d9938eafc9ae-0

@dobsonj dobsonj changed the title WIP: OCPBUGS-24061: Keep CSI operators progressing=true during DaemonSet rollout OCPBUGS-24061: Keep CSI operators progressing=true during DaemonSet rollout May 9, 2024
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label May 9, 2024
@openshift-ci

openshift-ci Bot commented May 9, 2024

Copy link
Copy Markdown
Contributor

@dobsonj: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-vsphere 412ef52 link true /test e2e-vsphere
ci/prow/e2e-vsphere-zones 412ef52 link false /test e2e-vsphere-zones
ci/prow/e2e-vsphere-csi-extended 412ef52 link false /test e2e-vsphere-csi-extended

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@dobsonj

dobsonj commented May 9, 2024

Copy link
Copy Markdown
Member Author

/retest

@dobsonj

dobsonj commented May 9, 2024

Copy link
Copy Markdown
Member Author

/payload-job periodic-ci-openshift-release-master-nightly-4.16-e2e-vsphere-ovn-techpreview-serial

@openshift-ci

openshift-ci Bot commented May 9, 2024

Copy link
Copy Markdown
Contributor

@dobsonj: trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-vsphere-ovn-techpreview-serial

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/0dd9be30-0e49-11ef-967a-5095acf95efd-0

@bertinatto bertinatto left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/label docs-approved
/label px-approved
This isn't a user-facing change.

@openshift-ci openshift-ci Bot added docs-approved Signifies that Docs has signed off on this PR px-approved Signifies that Product Support has signed off on this PR labels May 10, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 10, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 483de9c into openshift:master May 10, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dobsonj: Jira Issue OCPBUGS-24061: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-24061 has been moved to the MODIFIED state.

Details

In response to this:

https://issues.redhat.com/browse/OCPBUGS-24061

Bump library-go to get openshift/library-go#1734

go get github.com/openshift/library-go && go mod tidy && go mod vendor

/cc @openshift/storage

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-bot

Copy link
Copy Markdown

[ART PR BUILD NOTIFIER]

This PR has been included in build ose-vmware-vsphere-csi-driver-operator-container-v4.16.0-202405101842.p0.g483de9c.assembly.stream.el9 for distgit ose-vmware-vsphere-csi-driver-operator.
All builds following this will include this PR.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.16.0-0.nightly-2024-05-14-095225

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. docs-approved Signifies that Docs has signed off on this PR jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. px-approved Signifies that Product Support has signed off on this PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants