Skip to content

Allow override of BUILD_PLATFORM - #55

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
shiftstack:configurable-BUILD_PLATFORM
May 7, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
shiftstack:configurable-BUILD_PLATFORM

Conversation

@stephenfin

Copy link
Copy Markdown
Contributor

We don't need to build for all platforms if we already know what we're targeting.

We don't need to build for all platforms if we already know what we're
targeting.

Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
@openshift-ci
openshift-ci Bot requested review from rhmdnd and richardsonnick May 6, 2026 12:10
@openshift-ci

openshift-ci Bot commented May 6, 2026

Copy link
Copy Markdown
Contributor

@stephenfin: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/default-tls d672977 link false /test default-tls

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@damdo damdo left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

/assign @richardsonnick @smith-xyz

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 7, 2026
@richardsonnick

Copy link
Copy Markdown
Contributor

/lgtm

@smith-xyz

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented May 7, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: smith-xyz, stephenfin

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 7, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit eee0a3a into openshift:main May 7, 2026
6 of 7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants