Simplify rhel golang image publishing - #8656
Merged
openshift-merge-robot merged 1 commit intoApr 29, 2020
Merged
Conversation
jupierce
commented
Apr 29, 2020
Contributor
- Allow art's publish credentials to update builder-base and builder. As new rhel compilers are brought online, the ART job can tag these directly.
- Remove the origin-release build for rhel golang compilers. If the image can't use the rhel images directly, the images won't build with ART.
stevekuznetsov
approved these changes
Apr 29, 2020
jupierce
force-pushed
the
simplify_rhel_golang
branch
from
April 29, 2020 20:40
c74f9fd to
168bda7
Compare
Contributor
|
/lgtm |
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: gabemontero, jupierce, stevekuznetsov The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Member
|
This made the branch-ci-openshift-release-master-core-apply fail with: |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.