Skip to content

Revert 75520 - #76705

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
wewang58:revert_75520
Mar 23, 2026
Merged

Revert 75520#76705
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
wewang58:revert_75520

Conversation

@wewang58

@wewang58 wewang58 commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

Revert: #75520
Plan to add cilium test in e2e-aks.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 23, 2026
@openshift-ci

openshift-ci Bot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@wewang58
wewang58 marked this pull request as ready for review March 23, 2026 16:28
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 23, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@wewang58: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-hypershift-release-4.23-e2e-aks-cilium openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.22-e2e-aks-cilium openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-5.0-e2e-aks-cilium openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-main-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-main-e2e-aks-4-21 openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-main-e2e-aks-override openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-5.0-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-5.0-e2e-aks-4-21 openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-5.0-e2e-aks-override openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.23-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.23-e2e-aks-4-21 openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.23-e2e-aks-override openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-aks-4-21 openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-aks-override openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-aks-override openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.20-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.19-e2e-aks openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-azure-aks-external-oidc openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.20-e2e-azure-aks-external-oidc openshift/hypershift presubmit Registry content changed
pull-ci-openshift-cluster-ingress-operator-master-hypershift-e2e-aks openshift/cluster-ingress-operator presubmit Registry content changed
pull-ci-openshift-cluster-ingress-operator-release-5.0-hypershift-e2e-aks openshift/cluster-ingress-operator presubmit Registry content changed
pull-ci-openshift-cluster-ingress-operator-release-4.23-hypershift-e2e-aks openshift/cluster-ingress-operator presubmit Registry content changed
pull-ci-openshift-cluster-ingress-operator-release-4.22-hypershift-e2e-aks openshift/cluster-ingress-operator presubmit Registry content changed

A total of 89 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci
openshift-ci Bot requested review from bryan-cox and devguyio March 23, 2026 16:30
@bryan-cox

Copy link
Copy Markdown
Member

/approve

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 23, 2026
@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Mar 23, 2026
@wewang58 wewang58 closed this Mar 23, 2026
@wewang58 wewang58 reopened this Mar 23, 2026
@openshift-ci

openshift-ci Bot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

@wewang58: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@bryan-cox

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Mar 23, 2026
@openshift-ci

openshift-ci Bot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bryan-cox, wewang58

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 1053509 into openshift:main Mar 23, 2026
17 checks passed
BaiyangZhou pushed a commit to BaiyangZhou/release that referenced this pull request Mar 24, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
eliorerz pushed a commit to eliorerz/release that referenced this pull request Jun 3, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
fjglira pushed a commit to fjglira/release that referenced this pull request Jun 3, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
andrej1991 pushed a commit to andrej1991/release that referenced this pull request Jul 6, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
SachinNinganure pushed a commit to SachinNinganure/release that referenced this pull request Jul 20, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
malingatembo pushed a commit to malingatembo/release that referenced this pull request Aug 5, 2026
* Revert "Update hypershift azure e2e config and commands (openshift#75520)"

This reverts commit 394e5e0.

* Revert cilium job
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants