Skip to content

CORS-4314: gcp upi via infrastructure manager - #76144

Closed
jianli-wei wants to merge 1 commit into
openshift:mainfrom
jianli-wei:gcp-upi-infr-manager
Closed

jianli-wei wants to merge 1 commit into
openshift:mainfrom
jianli-wei:gcp-upi-infr-manager

Conversation

@jianli-wei

@jianli-wei jianli-wei commented Mar 12, 2026

Copy link
Copy Markdown
Contributor
  • FYI the installer PR CORS-4316: Update the Installer GCP UPI documentation. installer#10235
  • Add steps upi-install-gcp-by-infra-manager & upi-deprovision-gcp-by-infra-manager, for UPI using Infrastructure Manager
  • Update steps upi-install-gcp & upi-deprovision-gcp, for UPI using the pure "gcloud" commands (e.g. "gcloud compute...", "gcloud dns...", etc.)
  • Add chains upi-install-gcp & upi-deprovision-gcp, which include both the existing solution and the infra-manager solution which will be in use if infra-manager resources are available
  • FYI
    • the rehearsal test for UPI using Infrastructure Manager
    • the rehearsal test for UPI using the pure "gcloud" commands
  • TODOs: will do in another PR

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 12, 2026
@openshift-ci
openshift-ci Bot requested review from petr-muller and smg247 March 12, 2026 08:41
@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch 7 times, most recently from d4721e3 to d13b61d Compare March 13, 2026 02:47
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch from d13b61d to 55860ca Compare March 13, 2026 03:44
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch from 55860ca to ef71bdd Compare March 13, 2026 05:07
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch 2 times, most recently from 0a825eb to 79ed572 Compare March 13, 2026 06:16
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch 2 times, most recently from 9d33b39 to 59dcd4c Compare March 13, 2026 07:13
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch from 59dcd4c to ebe29ec Compare March 13, 2026 08:33
@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei
jianli-wei force-pushed the gcp-upi-infr-manager branch from ebe29ec to 4502918 Compare March 13, 2026 09:44
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

openshift-ci-robot commented Mar 31, 2026

Copy link
Copy Markdown
Contributor

@jianli-wei: This pull request references CORS-4314 which is a valid jira issue.

Details

In response to this:

  • FYI the installer PR CORS-4316: Update the Installer GCP UPI documentation. installer#10235
  • Add steps upi-install-gcp-by-infra-manager & upi-deprovision-gcp-by-infra-manager, for UPI using Infrastructure Manager
  • Update steps upi-install-gcp & upi-deprovision-gcp, for UPI using the pure "gcloud" commands (e.g. "gcloud compute...", "gcloud dns...", etc.)
  • Add chains upi-install-gcp & upi-deprovision-gcp, which include both the existing solution and the infra-manager solution which will be in use if infra-manager resources are available
  • FYI
  • the rehearsal test for UPI using Infrastructure Manager
  • the rehearsal test for UPI using the pure "gcloud" commands

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.22-installer-rehearse-gcp2

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@jianli-wei: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-cluster-etcd-operator-release-4.12-e2e-gcp-qe-no-capabilities openshift/cluster-etcd-operator presubmit Registry content changed
pull-ci-openshift-cluster-etcd-operator-release-4.11-e2e-gcp-qe-no-capabilities openshift/cluster-etcd-operator presubmit Registry content changed
pull-ci-openshift-installer-main-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-5.0-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.23-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.22-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.21-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.20-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.19-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.18-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.17-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.16-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.15-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.14-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.13-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.12-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.11-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.10-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.9-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.8-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.7-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.6-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-main-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-5.0-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.23-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed

A total of 167 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@jianli-wei: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-cluster-etcd-operator-release-4.12-e2e-gcp-qe-no-capabilities openshift/cluster-etcd-operator presubmit Registry content changed
pull-ci-openshift-cluster-etcd-operator-release-4.11-e2e-gcp-qe-no-capabilities openshift/cluster-etcd-operator presubmit Registry content changed
pull-ci-openshift-installer-main-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-5.0-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.23-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.22-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.21-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.20-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.19-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.18-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.17-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.16-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.15-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.14-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.13-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.12-e2e-gcp-ovn-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.11-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.10-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.9-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.8-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.7-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.6-e2e-gcp-upi openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-main-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-5.0-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.23-e2e-gcp-upi-xpn openshift/installer presubmit Registry content changed

A total of 167 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.22-installer-rehearse-gcp2

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.22-installer-rehearse-gcp2

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-verification-tests-main-installation-nightly-4.16-gcp-upi-xpn-f28

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

openshift-ci-robot commented Apr 3, 2026

Copy link
Copy Markdown
Contributor

@jianli-wei: This pull request references CORS-4314 which is a valid jira issue.

Details

In response to this:

  • FYI the installer PR CORS-4316: Update the Installer GCP UPI documentation. installer#10235
  • Add steps upi-install-gcp-by-infra-manager & upi-deprovision-gcp-by-infra-manager, for UPI using Infrastructure Manager
  • Update steps upi-install-gcp & upi-deprovision-gcp, for UPI using the pure "gcloud" commands (e.g. "gcloud compute...", "gcloud dns...", etc.)
  • Add chains upi-install-gcp & upi-deprovision-gcp, which include both the existing solution and the infra-manager solution which will be in use if infra-manager resources are available
  • FYI
  • the rehearsal test for UPI using Infrastructure Manager
  • the rehearsal test for UPI using the pure "gcloud" commands

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot

openshift-ci-robot commented Apr 3, 2026

Copy link
Copy Markdown
Contributor

@jianli-wei: This pull request references CORS-4314 which is a valid jira issue.

Details

In response to this:

  • FYI the installer PR CORS-4316: Update the Installer GCP UPI documentation. installer#10235
  • Add steps upi-install-gcp-by-infra-manager & upi-deprovision-gcp-by-infra-manager, for UPI using Infrastructure Manager
  • Update steps upi-install-gcp & upi-deprovision-gcp, for UPI using the pure "gcloud" commands (e.g. "gcloud compute...", "gcloud dns...", etc.)
  • Add chains upi-install-gcp & upi-deprovision-gcp, which include both the existing solution and the infra-manager solution which will be in use if infra-manager resources are available
  • FYI
  • the rehearsal test for UPI using Infrastructure Manager
  • the rehearsal test for UPI using the pure "gcloud" commands
  • TODOs: will do in another PR
  • Remove the Google Cloud SDK check, to use the configured one in upi-installer (see OCPBUGS-112027: update to Google Cloud SDK 563.0.0 installer#10460)
  • Update all gcp provision/deprovision steps, to use 4.22 upi-installer

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Apr 3, 2026

Copy link
Copy Markdown
Contributor

@jianli-wei: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/periodic-ci-openshift-verification-tests-main-installation-nightly-4.22-gcp-upi-xpn-f28 684dfcc link unknown /pj-rehearse periodic-ci-openshift-verification-tests-main-installation-nightly-4.22-gcp-upi-xpn-f28
ci/rehearse/periodic-ci-openshift-openshift-tests-private-release-4.22-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14 0163fe8 link unknown /pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.22-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14
ci/rehearse/periodic-ci-openshift-openshift-tests-private-release-4.23-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14 0163fe8 link unknown /pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.23-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14
ci/rehearse/periodic-ci-openshift-openshift-tests-private-release-4.21-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14 0163fe8 link unknown /pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.21-multi-nightly-gcp-upi-private-xpn-ingress-glb-arm-f14
ci/rehearse/periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp ee0c920 link unknown /pj-rehearse periodic-ci-openshift-verification-tests-main-installer-rehearse-4.23-installer-rehearse-gcp
ci/prow/generated-config 21be1e4 link true /test generated-config

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jianli-wei: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jianli-wei

Copy link
Copy Markdown
Contributor Author

@jianlinliu please help review, btw I'll deal with the TODOs in another PR because I'm not sure how long it'll take to get openshift/installer#10460 merged, thanks!

@barbacbd

Copy link
Copy Markdown
Contributor

/cc

@openshift-ci

openshift-ci Bot commented Apr 16, 2026

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@barbacbd

Copy link
Copy Markdown
Contributor

Closing in favor of #77934.

@barbacbd

Copy link
Copy Markdown
Contributor

/close

@openshift-ci

openshift-ci Bot commented Apr 16, 2026

Copy link
Copy Markdown
Contributor

@barbacbd: Closed this PR.

Details

In response to this:

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants