Skip to content

Debug - #72642

Closed
wewang58 wants to merge 3 commits into
openshift:mainfrom
wewang58:debug
Closed

Debug#72642
wewang58 wants to merge 3 commits into
openshift:mainfrom
wewang58:debug

Conversation

@wewang58

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Dec 16, 2025
@openshift-ci

openshift-ci Bot commented Dec 16, 2025

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Dec 18, 2025
@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Dec 18, 2025
@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: job(s): periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks either don't exist or were not found to be affected, and cannot be rehearsed

@openshift-merge-robot openshift-merge-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Dec 22, 2025
@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Dec 22, 2025
@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Jan 9, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Jan 9, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Jan 9, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@wewang58: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-cluster-image-registry-operator-main-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-5.0-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-4.23-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-4.22-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-4.21-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-4.20-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cluster-image-registry-operator-release-4.19-hypershift-e2e-aks openshift/cluster-image-registry-operator presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-main-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-5.0-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-4.23-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-4.22-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-4.21-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-4.20-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cloud-network-config-controller-release-4.19-hypershift-e2e-aks openshift/cloud-network-config-controller presubmit Registry content changed
pull-ci-openshift-cluster-machine-approver-main-e2e-hypershift-aks openshift/cluster-machine-approver presubmit Registry content changed
pull-ci-openshift-cluster-machine-approver-release-5.0-e2e-hypershift-aks openshift/cluster-machine-approver presubmit Registry content changed
pull-ci-openshift-cluster-machine-approver-release-4.23-e2e-hypershift-aks openshift/cluster-machine-approver presubmit Registry content changed
pull-ci-openshift-cluster-machine-approver-release-4.22-e2e-hypershift-aks openshift/cluster-machine-approver presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-main-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-5.0-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-4.23-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-4.22-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-4.21-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-4.20-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed
pull-ci-openshift-cluster-api-provider-azure-release-4.19-hypershift-e2e-aks openshift/cluster-api-provider-azure presubmit Registry content changed

A total of 84 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@wewang58

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Apr 8, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Apr 8, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@wewang58

wewang58 commented Apr 8, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wewang58: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci

openshift-ci Bot commented Apr 8, 2026

Copy link
Copy Markdown
Contributor

@wewang58: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks ac52164 link unknown /pj-rehearse periodic-ci-openshift-hypershift-release-4.22-periodics-e2e-aks

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@mgencur

mgencur commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

/close

Debugging PR.

1 similar comment
@mgencur

mgencur commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

/close

Debugging PR.

@openshift-ci openshift-ci Bot closed this Apr 9, 2026
@openshift-ci

openshift-ci Bot commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

@mgencur: Closed this PR.

Details

In response to this:

/close

Debugging PR.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci

openshift-ci Bot commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

@mgencur: Closed this PR.

Details

In response to this:

/close

Debugging PR.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants