Skip to content

test global-pull-secret in 4.20 - #71678

Closed
gaol wants to merge 2 commits into
openshift:masterfrom
gaol:global_ps_420
Closed

test global-pull-secret in 4.20#71678
gaol wants to merge 2 commits into
openshift:masterfrom
gaol:global_ps_420

Conversation

@gaol

@gaol gaol commented Nov 21, 2025

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Nov 21, 2025
@openshift-ci

openshift-ci Bot commented Nov 21, 2025

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@gaol

gaol commented Nov 21, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-mgmt-global-ps-f7 periodic-ci-openshift-hypershift-release-4.20-periodics-e2e-aws-ovn-conformance

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol

gaol commented Nov 21, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-mgmt-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol gaol closed this Nov 28, 2025
@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Nov 28, 2025
@gaol gaol reopened this Nov 28, 2025
@openshift-ci-robot openshift-ci-robot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Nov 28, 2025
@gaol

gaol commented Nov 28, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7 periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-mgmt-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol

gaol commented Nov 28, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol

gaol commented Nov 28, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol

gaol commented Nov 29, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@gaol

gaol commented Dec 1, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci

openshift-ci Bot commented Dec 1, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: gaol
Once this PR has been reviewed and has the lgtm label, please assign gpei, juzhao for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@gaol: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
periodic-ci-quay-quay-tests-master-quay-hypershift-quay-e2e-tests-quay313-hypershift-ocp419 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.17-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-openldap-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.17-amd64-nightly-aws-ipi-ovn-hypershift-guest-advanced-f14 N/A periodic Registry content changed
periodic-ci-openshift-insights-operator-release-4.18-hosted-hypershift-insights-operator-e2e-tests N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.16-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-guest-efs-f14-destructive N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.21-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-guest-efs-f14-destructive N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-guest-dynamic-dns-ondemand-advanced-f7 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-openldap-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-aws-ipi-ovn-hypershift-guest-longduration-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-amd64-nightly-aws-ipi-ovn-hypershift-guest-longduration-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.14-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-guest-f60 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.16-amd64-nightly-aws-ipi-ovn-hypershift-fips-mgmt-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f14 N/A periodic Registry content changed
periodic-ci-openshift-insights-operator-release-4.16-hosted-hypershift-insights-operator-e2e-tests N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.14-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-mgmt-f60 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-rollback-nightly-aws-ipi-ovn-hypershift-inplace-f28 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-guest-efs-f7 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-aws-ipi-ovn-hypershift-guest-dynamicdns-ondemand-advanced-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.21-amd64-nightly-aws-ipi-ovn-hypershift-arm-nodepool-guest-efs-f7 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-multi-nightly-aws-ipi-ovn-amd-hypershift-guest-mto-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.21-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-openldap-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-f999-compliance N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-amd64-nightly-4.18-upgrade-from-stable-4.17-aws-ipi-ovn-hypershift-replace-f14 N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-amd64-nightly-aws-ipi-ovn-hypershift-fips-guest-openldap-f14 N/A periodic Registry content changed

A total of 133 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@gaol

gaol commented Dec 1, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@gaol: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci

openshift-ci Bot commented Dec 1, 2025

Copy link
Copy Markdown
Contributor

@gaol: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-mgmt-global-ps-f7 f2d959e link unknown /pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-aws-ipi-ovn-hypershift-mgmt-global-ps-f7
ci/rehearse/periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7 37b65c6 link unknown /pj-rehearse periodic-ci-openshift-openshift-tests-private-release-4.20-amd64-nightly-azure-aks-hypershift-ephemeral-creds-guest-global-ps-f7

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@gaol gaol closed this Dec 18, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants