okd-scos: fix etcd job and add installer-etcd-artifact image - #51654
Conversation
|
[REHEARSALNOTIFIER]
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals. Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
|
/pj-rehearse pull-ci-openshift-etcd-openshift-4.16-okd-scos-images |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
| - name: TAGS | ||
| value: scos | ||
| dockerfile_path: Dockerfile.art | ||
| dockerfile_path: Dockerfile.rhel |
There was a problem hiding this comment.
@jupierce etcd uses Dockerfile.rhel in its other jobs.. should ocp-build-data be updated?
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hasbro17, Prashanth684 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/pj-rehearse ack |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
@Prashanth684: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in #50118. We are going through all images and identifying the ones which do not build and are working to fix them. This PR mimics the changes in the other configuration files and also adds the artifact image building.