okd-scos: fix machine-os-images job - #51605
Conversation
|
/pj-rehearse pull-ci-openshift-machine-os-images-main-okd-scos-images |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
/cc @jlebon |
|
@Prashanth684 Without any context provided in the commit message, I cannot review this. |
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them. This PR mimics the change in the `openshift-machine-os-images-main__okd.yaml` file to set the DIRECT_DOWNLOAD to be true.
c0c2a86 to
e211cee
Compare
sorry about that. added some context for the changes. |
|
[REHEARSALNOTIFIER]
Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
|
Thanks, looks sane to me. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jlebon, Prashanth684 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/pj-rehearse ack |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
@Prashanth684: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them. This PR mimics the change in the `openshift-machine-os-images-main__okd.yaml` file to set the DIRECT_DOWNLOAD to be true.
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in #50118. We are going through all images and identifying the ones which do not build and are working to fix them.
This PR mimics the change in the
openshift-machine-os-images-main__okd.yamlfile to set the DIRECT_DOWNLOAD to be true.