Skip to content

OKD-scos: fix hyperkube, pod build - #51601

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
Prashanth684:okd-pod
May 7, 2024
Merged

OKD-scos: fix hyperkube, pod build#51601
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
Prashanth684:okd-pod

Conversation

@Prashanth684

@Prashanth684 Prashanth684 commented May 2, 2024

Copy link
Copy Markdown
Contributor

In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in #50118. We are going through all images and identifying the ones which do not build and are working to fix them.This PR fixes the context for the dockerfile.

@openshift-ci
openshift-ci Bot requested review from deads2k and soltysh May 2, 2024 20:04
@Prashanth684

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-kubernetes-master-okd-scos-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@Prashanth684 Prashanth684 changed the title OKD-scos: fix pod build OKD-scos: fix hyperkube, pod build May 2, 2024
@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label May 6, 2024
@Prashanth684

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-kubernetes-master-okd-scos-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@Prashanth684

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-kubernetes-master-okd-scos-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: job(s): pull-ci-openshift-kubernetes-master-okd-scos-images either don't exist or were not found to be affected, and cannot be rehearsed

In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them.This PR fixes the context for the dockerfile.
@openshift-ci-robot openshift-ci-robot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label May 6, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@Prashanth684: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-kubernetes-master-okd-scos-images openshift/kubernetes presubmit Ci-operator config changed
Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@Prashanth684

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-kubernetes-master-okd-scos-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci

openshift-ci Bot commented May 6, 2024

Copy link
Copy Markdown
Contributor

@Prashanth684: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 7, 2024
@openshift-ci

openshift-ci Bot commented May 7, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Prashanth684, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 7, 2024
@Prashanth684

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label May 7, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 1a8f9d8 into openshift:master May 7, 2024
acornett21 pushed a commit to acornett21/openshift-release that referenced this pull request May 8, 2024
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them.This PR fixes the context for the dockerfile.
jbpratt pushed a commit to jbpratt/release that referenced this pull request May 30, 2024
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them.This PR fixes the context for the dockerfile.
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants