Skip to content

additional role for ci-staging ci-operator SA - #45160

Merged
openshift-ci[bot] merged 1 commit into
openshift:masterfrom
smg247:ci-stag-op-role-correction
Nov 1, 2023
Merged

additional role for ci-staging ci-operator SA#45160
openshift-ci[bot] merged 1 commit into
openshift:masterfrom
smg247:ci-stag-op-role-correction

Conversation

@smg247

@smg247 smg247 commented Nov 1, 2023

Copy link
Copy Markdown
Member

Correction for wrong name from #45159

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Nov 1, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@smg247: no rehearsable tests are affected by this change

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 1, 2023
@openshift-ci

openshift-ci Bot commented Nov 1, 2023

Copy link
Copy Markdown
Contributor

@smg247: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 1, 2023
@openshift-ci

openshift-ci Bot commented Nov 1, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: droslean, smg247

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci
openshift-ci Bot merged commit 201ac32 into openshift:master Nov 1, 2023
@smg247
smg247 deleted the ci-stag-op-role-correction branch August 9, 2024 18:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants