SPLAT-2794: Added 3CMO OTE binary to extensionBinary list - #31303
Conversation
|
Skipping CI for Draft Pull Request. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
✨ Finishing Touches🧪 Generate unit tests (beta)
Comment |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/20301940-68d2-11f1-85e8-271c25976bd8-0 |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/e9cb4960-68d6-11f1-9324-754438e212bb-0 |
|
/hold |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/6ae0e740-698f-11f1-9059-2e224b650e91-0 |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/fb401660-69be-11f1-954b-2036b46d4055-0 |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/2358e880-6a3b-11f1-9754-9355acf59ad9-0 |
|
@vr4manta: This PR was included in a payload test run from openshift/cluster-cloud-controller-manager-operator#471
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/497292d0-6b05-11f1-8161-670e9b7a11ab-0 |
|
Payload-job ran and succeeded above. In there we can see the two new e2e's run and completed successfully. |
|
@vr4manta Can you point me at where this work was done in 5.0 / main branch? |
Sure. |
|
/unhold |
|
/payload-job periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-hybrid-env-techpreview |
|
@vr4manta: trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command
See details on https://pr-payload-tests.ci.openshift.org/runs/ci/cac5bf70-7a22-11f1-8599-e0e1060309ce-0 |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: sosiouxme, vr4manta The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/label backport-risk-assessed |
|
/jira refresh |
|
@vr4manta: No Jira issue is referenced in the title of this pull request. Retaining the jira/valid-bug label as it was manually added. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@vr4manta: This pull request references SPLAT-2794 which is a valid jira issue. Retaining the jira/valid-bug label as it was manually added. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
lol a space before the semi-colon prevented bot from detecting Jira. |
|
/verified by @vr4manta |
|
@vr4manta: This PR has been marked as verified by DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
47dc47b
into
openshift:release-4.22
SPLAT-2794
Changes
Prereqs
Notes
Original Main branch (5.0) PRs: