Skip to content

OTA-1174: upgrade status: add --detailed=nodes that shows all nodes - #1735

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
petr-muller:ota-1174-details-nodes
Apr 18, 2024
Merged

OTA-1174: upgrade status: add --detailed=nodes that shows all nodes#1735
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
petr-muller:ota-1174-details-nodes

Conversation

@petr-muller

@petr-muller petr-muller commented Apr 17, 2024

Copy link
Copy Markdown
Member

Deviate from the mockup and make --detailed accept an enum value instead of being a boolean. This allows expanding different sections which can be useful in different scenarios.

For now, allow the following values:

  • none: non-detailed output, default
  • nodes: nodes sections are not capped to 10 nodes
  • health: health insights provide further details (not implemented yet)
  • all: all sections capable to emit detailed output do so

Also, test --detailed=all in the examples test.

Deviate from the mockup and make `--detailed` accept an enum value
instead of being a boolean. This allows expanding different sections
which can be useful in different scenarios.

For now, allow the following values:
- `none`: non-detailed output, default
- `nodes`: nodes sections are not capped to 10 nodes
- `health`: health insights provide further details (not implemented yet)
- `all`: all sections capable to emit detailed output do so
@petr-muller petr-muller changed the title upgrade status: add --detailed=nodes that shows all nodes OTA-1174: upgrade status: add --detailed=nodes that shows all nodes Apr 17, 2024
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 17, 2024
@openshift-ci-robot

openshift-ci-robot commented Apr 17, 2024

Copy link
Copy Markdown

@petr-muller: This pull request references OTA-1174 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.16.0" version, but no target version was set.

Details

In response to this:

Deviate from the mockup and make --detailed accept an enum value instead of being a boolean. This allows expanding different sections which can be useful in different scenarios.

For now, allow the following values:

  • none: non-detailed output, default
  • nodes: nodes sections are not capped to 10 nodes
  • health: health insights provide further details (not implemented yet)
  • all: all sections capable to emit detailed output do so

Also, test --detailed=all in the examples test.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

build0-gstfj-w-b-infra-n9kc5.c.openshift-ci-build-farm.internal Completed Updated 4.16.0-ec.3 -
build0-gstfj-w-b-mwf8t.c.openshift-ci-build-farm.internal Completed Updated 4.16.0-ec.3 -
build0-gstfj-w-c-nsmn5.c.openshift-ci-build-farm.internal Completed Updated 4.16.0-ec.3 -
build0-gstfj-w-c-qjbdl.c.openshift-ci-build-farm.internal Completed Updated 4.16.0-ec.3 -

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is the only output that differs from its non-detailed variant (for now)

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is good. When we have a detailed output how do we tell the user where to focus or what is the main toot cause of these issues. Because I am familier with OCP I can guess that the issue is

 Reason: ClusterOperatorDegraded
  Message: Cluster operator machine-config is degraded

But will that be obvious to the user as well?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Hopefully the remaining bits from #1690 (--details=health or --details=all) will help with that, expanding the = Update Health = section items with "description", "reference" and "resources" fields. That PR grew too much so I'm breaking it down to smaller, easier ones. This PR just brings the change to implement turning off the omission:

Omitted additional 49 Total, 22 Completed, 46 Available, 3 Progressing, 27 Outdated, 3 Draining, 0 Excluded, and 0 Degraded nodes.
Pass along --details=nodes to see all information.

I'm convinced that "I want to see all nodes" and "I want to see details about health insights" are separate use cases so I split them this way.

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Check out where #1690 eventually takes this:

Message: 7 nodes in pool worker are degraded while draining: failed to drain node: <name> after 1 hour. Please see machine-config-controller logs for more information
Since: -
Level: Error
Impact: Update Stalled
Reference: https://docs.openshift.com/container-platform/4.15/post_installation_configuration/machine-configuration-tasks.html#understanding-the-machine-config-operator
Resources:
Nodes: build0-gstfj-ci-prowjobs-worker-b-9lztv build0-gstfj-ci-prowjobs-worker-b-bg9f5 build0-gstfj-ci-prowjobs-worker-b-mrxwn build0-gstfj-ci-tests-worker-b-4h7pn build0-gstfj-ci-tests-worker-b-jv5bg build0-gstfj-ci-tests-worker-b-kj6gk build0-gstfj-ci-tests-worker-c-dcz9p
Description: failed to drain node: <name> after 1 hour. Please see machine-config-controller logs for more information

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@petr-muller sounds good to me.

@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik April 17, 2024 14:07
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 17, 2024
@petr-muller

Copy link
Copy Markdown
Member Author

/retest

@petr-muller

Copy link
Copy Markdown
Member Author
level=error msg=Error: Unable to list provider registration status, it is possible that this is due to invalid credentials or the service principal does not have permission to use the Resource Manager API, Azure error: resources.ProvidersClient#List: Failure responding to request: StatusCode=200 -- Original Error: Error occurred reading http.Response#Body - Error = 'read tcp 10.130.46.72:56638->4.150.241.10:443: read: connection reset by peer'

Looks unrelated

@petr-muller

Copy link
Copy Markdown
Member Author

/retest

@LalatenduMohanty LalatenduMohanty left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 18, 2024
@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: LalatenduMohanty, petr-muller

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

@petr-muller: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-metal-ipi-ovn-ipv6 38cd6c9 link false /test e2e-metal-ipi-ovn-ipv6

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit b3dfe36 into openshift:master Apr 18, 2024
@petr-muller
petr-muller deleted the ota-1174-details-nodes branch April 18, 2024 23:04
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202404190512.p0.gb3dfe36.assembly.stream.el8 for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants