Skip to content

Fix build error from missed unit tests refactoring - #939

Merged
kaovilai merged 1 commit into
openshift:masterfrom
shubham-pampattiwar:fix-install
Mar 27, 2023
Merged

kaovilai merged 1 commit into
openshift:masterfrom
shubham-pampattiwar:fix-install

Conversation

@shubham-pampattiwar

Copy link
Copy Markdown
Member

No description provided.

@openshift-ci
openshift-ci Bot requested review from eemcmullan and sseago March 27, 2023 13:53
@kaovilai

Copy link
Copy Markdown
Member

/override ci/prow/4.11-operator-e2e-gcp

Failed to destroy cluster: failed to cache machine types: googleapi: Error 403: Quota exceeded for quota metric 'Heavy-weight read requests' and limit 'Heavy-weight read requests per minute' of service 'compute.googleapis.com' for consumer 'project_number:1053217076791'.
level=fatal msg=Details:
level=fatal msg=[
level=fatal msg=  {
level=fatal msg=    "@type": "type.googleapis.com/google.rpc.ErrorInfo",
level=fatal msg=    "domain": "googleapis.com",
level=fatal msg=    "metadatas": {
level=fatal msg=      "consumer": "projects/1053217076791",
level=fatal msg=      "quota_limit": "HeavyWeightReadRequestsPerMinutePerProject",
level=fatal msg=      "quota_limit_value": "750",
level=fatal msg=      "quota_location": "global",
level=fatal msg=      "quota_metric": "compute.googleapis.com/heavy_weight_read_requests",
level=fatal msg=      "service": "compute.googleapis.com"
level=fatal msg=    },
level=fatal msg=    "reason": "RATE_LIMIT_EXCEEDED"
level=fatal msg=  },
level=fatal msg=  {
level=fatal msg=    "@type": "type.googleapis.com/google.rpc.Help",
level=fatal msg=    "links": [
level=fatal msg=      {
level=fatal msg=        "description": "Request a higher quota limit.",
level=fatal msg=        "url": "https://cloud.google.com/docs/quota#requesting_higher_quota"
level=fatal msg=      }
level=fatal msg=    ]
level=fatal msg=  }
level=fatal msg=]
level=fatal msg=, rateLimitExceeded

@openshift-ci

openshift-ci Bot commented Mar 27, 2023

Copy link
Copy Markdown

@kaovilai: Overrode contexts on behalf of kaovilai: ci/prow/4.11-operator-e2e-gcp

Details

In response to this:

/override ci/prow/4.11-operator-e2e-gcp

Failed to destroy cluster: failed to cache machine types: googleapi: Error 403: Quota exceeded for quota metric 'Heavy-weight read requests' and limit 'Heavy-weight read requests per minute' of service 'compute.googleapis.com' for consumer 'project_number:1053217076791'.
level=fatal msg=Details:
level=fatal msg=[
level=fatal msg=  {
level=fatal msg=    "@type": "type.googleapis.com/google.rpc.ErrorInfo",
level=fatal msg=    "domain": "googleapis.com",
level=fatal msg=    "metadatas": {
level=fatal msg=      "consumer": "projects/1053217076791",
level=fatal msg=      "quota_limit": "HeavyWeightReadRequestsPerMinutePerProject",
level=fatal msg=      "quota_limit_value": "750",
level=fatal msg=      "quota_location": "global",
level=fatal msg=      "quota_metric": "compute.googleapis.com/heavy_weight_read_requests",
level=fatal msg=      "service": "compute.googleapis.com"
level=fatal msg=    },
level=fatal msg=    "reason": "RATE_LIMIT_EXCEEDED"
level=fatal msg=  },
level=fatal msg=  {
level=fatal msg=    "@type": "type.googleapis.com/google.rpc.Help",
level=fatal msg=    "links": [
level=fatal msg=      {
level=fatal msg=        "description": "Request a higher quota limit.",
level=fatal msg=        "url": "https://cloud.google.com/docs/quota#requesting_higher_quota"
level=fatal msg=      }
level=fatal msg=    ]
level=fatal msg=  }
level=fatal msg=]
level=fatal msg=, rateLimitExceeded

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@kaovilai

Copy link
Copy Markdown
Member

/test 4.8-operator-e2e-aws
/test 4.9-operator-e2e-aws

@kaovilai kaovilai changed the title Fix tests missed during refactoring Fix build error from missed unit tests refactoring Mar 27, 2023
@sseago

sseago commented Mar 27, 2023

Copy link
Copy Markdown
Contributor

/retest

@kaovilai

Copy link
Copy Markdown
Member

ci failed to launch cluster again. The issue we hit was build issue. I think it's ok to
/override ci/prow/4.9-operator-e2e-aws

@openshift-ci

openshift-ci Bot commented Mar 27, 2023

Copy link
Copy Markdown

@kaovilai: Overrode contexts on behalf of kaovilai: ci/prow/4.9-operator-e2e-aws

Details

In response to this:

ci failed to launch cluster again. The issue we hit was build issue. I think it's ok to
/override ci/prow/4.9-operator-e2e-aws

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@kaovilai
kaovilai merged commit d1c95dc into openshift:master Mar 27, 2023
@openshift-ci

openshift-ci Bot commented Mar 27, 2023

Copy link
Copy Markdown

@shubham-pampattiwar: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants