Skip to content

[OADP-458] Bring back component: velero label - #638

Merged
kaovilai merged 2 commits into
openshift:masterfrom
kaovilai:labels
Apr 18, 2022
Merged

kaovilai merged 2 commits into
openshift:masterfrom
kaovilai:labels

Conversation

@kaovilai

@kaovilai kaovilai commented Apr 13, 2022

Copy link
Copy Markdown
Member

fix OADP-458

@JaydipGabani

JaydipGabani commented Apr 13, 2022

Copy link
Copy Markdown
Contributor

@kaovilai I am a little confused so bear with me, the label we are looking for is component: velero which I don't see in these changes, unless that is somehow a default label that now is going to be present. Would you be able to post velero pod yaml in your PR here to display that the desired label is now present after your changes?

Comment thread controllers/velero.go
)
veleroDeployment.TypeMeta = installDeployment.TypeMeta
veleroDeployment.Spec = installDeployment.Spec
veleroDeployment.Labels = installDeployment.Labels

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@JaydipGabani This adds component: velero to veleroDeployment

Comment thread controllers/velero.go
}

func (r *DPAReconciler) customizeVeleroDeployment(dpa *oadpv1alpha1.DataProtectionApplication, veleroDeployment *appsv1.Deployment) error {
veleroDeployment.Labels = r.getAppLabels(dpa)

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We don't override labels

Comment thread controllers/velero.go
},
//append dpa labels
for k, v := range r.getDpaAppLabels(dpa) {
if veleroDeployment.Labels[k] == "" { //saves component: velero labels

@kaovilai kaovilai Apr 13, 2022

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Comment thread controllers/restic.go
ds.TypeMeta = installDs.TypeMeta
// Update Spec
ds.Spec = installDs.Spec
ds.Labels = installDs.Labels

@kaovilai kaovilai Apr 13, 2022

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@JaydipGabani this sets label for restic.

@kaovilai

Copy link
Copy Markdown
Member Author

/retest

@kaovilai

Copy link
Copy Markdown
Member Author

/retest

1 similar comment
@kaovilai

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented Apr 14, 2022

Copy link
Copy Markdown

@kaovilai: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@kaovilai
kaovilai merged commit a29c162 into openshift:master Apr 18, 2022
@kaovilai kaovilai changed the title Bring back component: velero label [OADP-458] Bring back component: velero label May 6, 2022
kaovilai added a commit to kaovilai/oadp-operator that referenced this pull request May 9, 2022
* Bring back `component: velero` label, fix OADP-452

* fix component:server on deployment, unit test update
dymurray pushed a commit that referenced this pull request May 9, 2022
#638) (#677)

* Bring back `component: velero` label (#638)

* Bring back `component: velero` label, fix OADP-452

* fix component:server on deployment, unit test update

* label diff for "given valid DPA CR, appropriate velero deployment is build with aws plugin specific specs"
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants