Update upstream kubevirt tag in manager deployment env, preserve #611 changes in config - #615
Merged
Merged
Conversation
Member
Author
|
/retest |
3 similar comments
Member
Author
|
/retest |
Member
Author
|
/retest |
Member
Author
|
/retest |
dymurray
approved these changes
Mar 31, 2022
dymurray
reviewed
Mar 31, 2022
| operatorframework.io/suggested-namespace: openshift-adp | ||
| operators.openshift.io/infrastructure-features: '["Disconnected"]' | ||
| operators.openshift.io/valid-subscription: '["OpenShift Container Platform", "OpenShift | ||
| Platform Plus"]' |
Member
Author
There was a problem hiding this comment.
dymurray
approved these changes
Apr 7, 2022
|
@kaovilai: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
kaovilai
added a commit
to kaovilai/oadp-operator
that referenced
this pull request
May 6, 2022
…shift#611 changes in config (openshift#615) * Preserve openshift#611 bundle changes in config manifests * prefix `v` to kubevirt tag
kaovilai
added a commit
to kaovilai/oadp-operator
that referenced
this pull request
May 6, 2022
…shift#611 changes in config (openshift#615) * Preserve openshift#611 bundle changes in config manifests * prefix `v` to kubevirt tag
dymurray
pushed a commit
that referenced
this pull request
May 9, 2022
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
preserve OADP-380 in config