Skip to content

Deprecate EtcdDiscoveryDomain field on ControllerConfig CRD - #1675

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
LorbusChris:etcd-discovery-domain
Jun 4, 2020
Merged

Deprecate EtcdDiscoveryDomain field on ControllerConfig CRD#1675
openshift-merge-robot merged 1 commit into
openshift:masterfrom
LorbusChris:etcd-discovery-domain

Conversation

@LorbusChris

Copy link
Copy Markdown
Contributor

Currently ControllerConfig.EtcdDiscoveryDomain and
ControllerConfig.Infra.Status.EtcdDiscoveryDomain are redundant..

- What I did
This deprecates the ControllerConfig.EtcdDiscoveryDomain field and
leaves ControllerConfig.Infra.Status.EtcdDiscoveryDomain as the
canonical location.

- How to verify it
CI

- Description for the changelog
Deprecate EtcdDiscvoeryDomain field on ControllerConfig CRD

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

/assign hexfusion

@yuqi-zhang

Copy link
Copy Markdown
Contributor

Looks sane. Could you update the typo in the commit message/title EtcdDiscvoeryDomain -> EtcdDiscoveryDomain

/test e2e-vsphere
/test e2e-openstack

@LorbusChris
LorbusChris force-pushed the etcd-discovery-domain branch from 61c11d4 to 5cc8813 Compare April 23, 2020 21:23
@LorbusChris LorbusChris changed the title Deprecate EtcdDiscvoeryDomain field on ControllerConfig CRD Deprecate EtcdDiscoveryDomain field on ControllerConfig CRD Apr 23, 2020
@LorbusChris

Copy link
Copy Markdown
Contributor Author

whoopsie. fixed :)

@LorbusChris

LorbusChris commented Apr 23, 2020

Copy link
Copy Markdown
Contributor Author

Also note that https://github.com/openshift/machine-config-operator/blob/master/pkg/operator/render.go#L98 currently isn't touched by this, so the field is still going to be populated.

Maybe I should add a comment there noting that it is deprecated and can eventually be removed one day?

Edit: Or would you rather have this removed from the rendered spec already?

@LorbusChris

Copy link
Copy Markdown
Contributor Author

/retest

@runcom

runcom commented Apr 24, 2020

Copy link
Copy Markdown
Member

Maybe I should add a comment there noting that it is deprecated and can eventually be removed one day?

I'm in favor of adding a comment saying it's unused for at least a release (even if we're really the one using that) - no biggie, we'll remove it afterward, what matters here is that we turned to use the infra one which is the new and correct one 👍

@runcom

runcom commented Apr 24, 2020

Copy link
Copy Markdown
Member

Will likely go in in 4.6 as it's not a bug, the valid bug will prevent this from merging
I'll leave the lgtm to the team once the comment is added

/approve

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 24, 2020
@LorbusChris
LorbusChris force-pushed the etcd-discovery-domain branch from 5cc8813 to 752338a Compare April 24, 2020 10:23
@LorbusChris

LorbusChris commented Apr 24, 2020

Copy link
Copy Markdown
Contributor Author

Added the comment

@LorbusChris

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp-op

@yuqi-zhang yuqi-zhang left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/hold until 4.6 just in case the BZ field slips for some reason

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 24, 2020
@openshift-ci-robot openshift-ci-robot added lgtm Indicates that a PR is ready to be merged. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. labels Apr 24, 2020
Currently ControllerConfig.EtcdDiscoveryDomain and
ControllerConfig.Infra.Status.EtcdDiscoveryDomain are redundant.

This deprecates the ControllerConfig.EtcdDiscoveryDomain field and
leaves ControllerConfig.Infra.Status.EtcdDiscoveryDomain as the
canonical location.
@LorbusChris
LorbusChris force-pushed the etcd-discovery-domain branch from 752338a to c3bf0fd Compare June 4, 2020 13:01
@openshift-ci-robot openshift-ci-robot removed lgtm Indicates that a PR is ready to be merged. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. labels Jun 4, 2020
@LorbusChris

Copy link
Copy Markdown
Contributor Author

PR rebased.

/hold cancel
/cc @yuqi-zhang @runcom

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jun 4, 2020
@runcom

runcom commented Jun 4, 2020

Copy link
Copy Markdown
Member

/approve
/lgtm

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: LorbusChris, runcom, yuqi-zhang

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jun 4, 2020
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@LorbusChris

Copy link
Copy Markdown
Contributor Author

/test e2e-metal-ipi

@openshift-ci-robot

openshift-ci-robot commented Jun 4, 2020

Copy link
Copy Markdown
Contributor

@LorbusChris: The following tests failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/e2e-openstack 61c11d454661eed1372af9e9cab990460c8e27a1 link /test e2e-openstack
ci/prow/e2e-vsphere 61c11d454661eed1372af9e9cab990460c8e27a1 link /test e2e-vsphere

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit f751a6b into openshift:master Jun 4, 2020
mandre added a commit to mandre/machine-config-operator that referenced this pull request Jun 23, 2020
The EtcdDiscoveryDomain field on ControllerConfig CRD was deprecated in
[1] in favor of `ControllerConfig.Infra.Status.EtcdDiscoveryDomain`.
However the originial PR missed two matches.

This fixes it so that there is no surprise in case
`ControllerConfig.EtcdDiscoveryDomain` is effectively removed.

[1] openshift#1675
ptalgulk01 pushed a commit to ptalgulk01/machine-config-operator that referenced this pull request May 15, 2026
Deprecate EtcdDiscoveryDomain field on ControllerConfig CRD
ptalgulk01 pushed a commit to ptalgulk01/machine-config-operator that referenced this pull request May 15, 2026
The EtcdDiscoveryDomain field on ControllerConfig CRD was deprecated in
[1] in favor of `ControllerConfig.Infra.Status.EtcdDiscoveryDomain`.
However the originial PR missed two matches.

This fixes it so that there is no surprise in case
`ControllerConfig.EtcdDiscoveryDomain` is effectively removed.

[1] openshift#1675
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

8 participants