Bug 1788522: config: default bind network to tcp instead of tcp4 [release-4.3] - #658
Conversation
|
@openshift-cherrypick-robot: No Bugzilla bug is referenced in the title of this pull request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@openshift-cherrypick-robot: This pull request references Bugzilla bug 1788522, which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/bugzilla refresh |
|
@mfojtik: This pull request references Bugzilla bug 1788522, which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/bugzilla refresh |
|
@mfojtik: This pull request references Bugzilla bug 1788522, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/bugzilla refresh |
|
@mfojtik: This pull request references Bugzilla bug 1788522, which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@eparis process note... Having a bug with target release (also moving this to 4.3.0 means we are going to create blocker that is not really a blocker ;-) |
|
For patch-manager: This is high priority fix, without this we can't bump operators (which will probably take another z-stream) and it means that operators in ipv6 clusters are not reporting metrics and healthz at all (because the HTTP server that does that won't bind on IPv6 interface). |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: openshift-cherrypick-robot, sttts The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
Overriding the labels manually per slack discussion with Eric. We are going to special-case library-go repository and it won't be necessary to have valid-bug label and cherry-pick approved for backports here. |
|
@openshift-cherrypick-robot: All pull requests linked via external trackers have merged. Bugzilla bug 1788522 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
This is an automated cherry-pick of #657
/assign mfojtik