Skip to content

fileobserver: prometheus counter for observed actions - #654

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
mfojtik:observer-metrics
Jan 3, 2020
Merged

fileobserver: prometheus counter for observed actions#654
openshift-merge-robot merged 1 commit into
openshift:masterfrom
mfojtik:observer-metrics

Conversation

@mfojtik

@mfojtik mfojtik commented Jan 3, 2020

Copy link
Copy Markdown
Contributor

This add fileobserver_action_count{name=ACTION, filename="file"}=N prometheus metric that can be used to track changes on particular files over time and indicate problems when files are changed too often or not changed at all.

@openshift-ci-robot openshift-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Jan 3, 2020
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 3, 2020

@soltysh soltysh left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jan 3, 2020
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mfojtik, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-bot

Copy link
Copy Markdown

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-ci-robot openshift-ci-robot removed the lgtm Indicates that a PR is ready to be merged. label Jan 3, 2020
@openshift-ci-robot

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@mfojtik mfojtik added the lgtm Indicates that a PR is ready to be merged. label Jan 3, 2020
@mfojtik

mfojtik commented Jan 3, 2020

Copy link
Copy Markdown
Contributor Author

(fixed build error, re-applying lgtm)

@openshift-merge-robot
openshift-merge-robot merged commit cd8a585 into openshift:master Jan 3, 2020
mfojtik added a commit to mfojtik/cluster-kube-controller-manager-operator that referenced this pull request Jan 7, 2020
openshift/library-go#657 : config: default bind network to tcp instead of tcp4
openshift/library-go#656 : leaderelection: set release on cancel to true
openshift/library-go#654 : fileobserver: prometheus counter for observed actions
mfojtik added a commit to mfojtik/cluster-openshift-apiserver-operator that referenced this pull request Jan 7, 2020
openshift/library-go#657 : config: default bind network to tcp instead of tcp4
openshift/library-go#656 : leaderelection: set release on cancel to true
openshift/library-go#654 : fileobserver: prometheus counter for observed actions
bertinatto pushed a commit to bertinatto/library-go that referenced this pull request Jul 2, 2020
fileobserver: prometheus counter for observed actions
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants