Skip to content

OCPBUGS-74511: remove RouteExternalCertificate feature gate - #2122

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
jcmoraisjr:OCPBUGS-74511-remove-featuregate
May 20, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
jcmoraisjr:OCPBUGS-74511-remove-featuregate

Conversation

@jcmoraisjr

Copy link
Copy Markdown
Member

RouteExternalCertificate featuregate is now enabled by default. This update is removing the opts used to hold its value, hardcoding the behavior when the value is true.

This update needs to be in sync with o/kubernetes and o/openshift-apiserver.

RouteExternalCertificate is now enabled by default. This update is
removing the opts used to hold its value, hardcoding the behavior when
the value is true.

This update needs to be in sync with o/kubernetes and
o/openshift-apiserver
@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Feb 10, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@jcmoraisjr: This pull request references Jira Issue OCPBUGS-74511, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.22.0) matches configured target version for branch (4.22.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

Requesting review from QA contact:
/cc @melvinjoseph86

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

RouteExternalCertificate featuregate is now enabled by default. This update is removing the opts used to hold its value, hardcoding the behavior when the value is true.

This update needs to be in sync with o/kubernetes and o/openshift-apiserver.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jcmoraisjr

jcmoraisjr commented Feb 11, 2026

Copy link
Copy Markdown
Member Author

/hold

needs to be merged after openshift/kubernetes#2585 and openshift/openshift-apiserver#604

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 11, 2026
@rikatz

rikatz commented Feb 25, 2026

Copy link
Copy Markdown
Member

/assign
/cc

@openshift-ci
openshift-ci Bot requested a review from rikatz February 25, 2026 15:50
@rikatz

rikatz commented Feb 27, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve
Thanks

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 27, 2026
@rikatz

rikatz commented Feb 27, 2026

Copy link
Copy Markdown
Member

/cc @bertinatto @p0lyn0mial

@bertinatto bertinatto left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

Is the feature enabled by default since 4.19?

@openshift-ci

openshift-ci Bot commented Mar 2, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bertinatto, jcmoraisjr, rikatz

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 2, 2026
@alebedev87

alebedev87 commented Mar 2, 2026

Copy link
Copy Markdown
Contributor

Is the feature enabled by default since 4.19?

@bertinatto: yes.

@jcmoraisjr

Copy link
Copy Markdown
Member Author

@bertinatto this change should be immediately followed by changes in both o/kubernetes and o/openshift-apiserver due to the removal of the whole struct. Can you support us in a way we bump and merge all of them quickly, so users trying to bump library-go are not impacted? Otherwise we can change the approach and instead make smaller cleanups on o/kubernetes and o/openshift-apiserver before changing library-go.

There is also a new comment here. Any preferable approach?

@bertinatto

Copy link
Copy Markdown
Member

@bertinatto this change should be immediately followed by changes in both o/kubernetes and o/openshift-apiserver due to the removal of the whole struct. Can you support us in a way we bump and merge all of them quickly, so users trying to bump library-go are not impacted? Otherwise we can change the approach and instead make smaller cleanups on o/kubernetes and o/openshift-apiserver before changing library-go.

There is also a new comment here. Any preferable approach?

It looks like you want to merge the library-go PR first, then bump it in o/k and o/openshift-apiserver.

The o/k and o/openshift-apiserver PRs can have:

  1. A commit bumping library-go (o/k has a special script for that, check the README)
  2. Any other commits you may need to drop the feature (including dead references to library-go code that you're removing)

I recommend you wait for @jacobsee and @jubittajohn to land the kube 1.35 update first.

I can approve PRs in o/k and o/library-go. You'll need find an approver for o/openshift-apiserver.

@melvinjoseph86

Copy link
Copy Markdown

/verified later @mjoseph

@openshift-ci-robot

Copy link
Copy Markdown

@melvinjoseph86: This PR has been marked to be verified later by @mjoseph.

Details

In response to this:

/verified later @mjoseph

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Mar 12, 2026
@openshift-bot

Copy link
Copy Markdown

/jira refresh

The requirements for Jira bugs have changed (Jira issues linked to PRs on main branch need to target different OCP), recalculating validity.

@openshift-ci-robot

Copy link
Copy Markdown

@openshift-bot: This pull request references Jira Issue OCPBUGS-74511, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.22.0) matches configured target version for branch (4.22.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

Requesting review from QA contact:
/cc @melvinjoseph86

Details

In response to this:

/jira refresh

The requirements for Jira bugs have changed (Jira issues linked to PRs on main branch need to target different OCP), recalculating validity.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jcmoraisjr

Copy link
Copy Markdown
Member Author

@bertinatto o/openshift-apiserver is still on k 1.34, but I missed the hold tag and it ended up approved and merged. I think I cannot update library-go there due to kube version mismatch, so folks should handle it when bumping kube libs.

I'm planning to unhold this one now, immediately followed by bumping its version on o/kubernetes and merging there as well. Sounds good, should we proceed?

@jcmoraisjr

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 20, 2026
@openshift-ci

openshift-ci Bot commented May 20, 2026

Copy link
Copy Markdown
Contributor

@jcmoraisjr: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 3a6f949 into openshift:master May 20, 2026
4 checks passed
@openshift-ci-robot

Copy link
Copy Markdown

@jcmoraisjr: Jira Issue OCPBUGS-74511: Some pull requests linked via external trackers have merged:

The following pull request, linked via external tracker, has not merged:

All associated pull requests must be merged or unlinked from the Jira bug in order for it to move to the next state. Once unlinked, request a bug refresh with /jira refresh.

Jira Issue OCPBUGS-74511 has not been moved to the MODIFIED state.

This PR is marked as verified-later. Jira issue(s) in the title of this PR will require post-merge verification. After testing, it must be manually moved to the VERIFIED state.

Details

In response to this:

RouteExternalCertificate featuregate is now enabled by default. This update is removing the opts used to hold its value, hardcoding the behavior when the value is true.

This update needs to be in sync with o/kubernetes and o/openshift-apiserver.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jcmoraisjr
jcmoraisjr deleted the OCPBUGS-74511-remove-featuregate branch May 20, 2026 18:08
gangwgr pushed a commit to gangwgr/library-go that referenced this pull request May 21, 2026
…e-featuregate

OCPBUGS-74511: remove RouteExternalCertificate feature gate
@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in release 5.0.0-0.nightly-2026-05-27-134409

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in release 5.0.0-0.nightly-2026-08-06-014706

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria verified-later

Projects

None yet

Development

Successfully merging this pull request may close these issues.

8 participants