Skip to content

crypto: add DefaultTLSProfileType const - #2111

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
damdo:add-default-tlsprofiletype
Feb 3, 2026
Merged

crypto: add DefaultTLSProfileType const#2111
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
damdo:add-default-tlsprofiletype

Conversation

@damdo

@damdo damdo commented Feb 3, 2026

Copy link
Copy Markdown
Member

Rather than relying on implicit knowledge that the configv1.TLSProfileIntermediateType is the default, we set a const that defines this, and that can be referenced also from consumers of this package (e.g. tls config oberver, openshift/controller-runtime-commons)

Context: it was originally discussed here: openshift/controller-runtime-common#2 (comment)

Rather than relying on implicit knowledge that the configv1.TLSProfileIntermediateType is the default, we set a const that defines this, and that can be referenced also from consumers of this package (e.g. tls config oberver, openshift/controller-runtime-commons)
@damdo

damdo commented Feb 3, 2026

Copy link
Copy Markdown
Member Author

/assign @joelanford @p0lyn0mial @bertinatto

@openshift-ci

openshift-ci Bot commented Feb 3, 2026

Copy link
Copy Markdown
Contributor

@damdo: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@joelanford

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 3, 2026

@bertinatto bertinatto left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@openshift-ci

openshift-ci Bot commented Feb 3, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bertinatto, damdo, joelanford

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 3, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 3067285 into openshift:master Feb 3, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants